)]}'
{"id":"openstack%2Fkolla~998266","triplet_id":"openstack%2Fkolla~master~I37bb0b5056b2e055d0d0c9da5d0beba4b9ba9067","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-07-22 07:09:38.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I37bb0b5056b2e055d0d0c9da5d0beba4b9ba9067","subject":"CI: copr - replace secrets without newlines","status":"MERGED","created":"2026-07-22 06:22:33.000000000","updated":"2026-07-22 07:26:04.000000000","submitted":"2026-07-22 07:24:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998266","meta_rev_id":"45a11cc7fbc2615060a62d64fec274635c193071","_number":998266,"virtual_id_number":998266,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-22 07:24:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-07-22 07:09:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-07-22 07:09:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-07-22 07:09:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-22 06:29:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-22 06:35:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d06a9dae463519e47a56e3bf3a7cc865058e81e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-22 06:22:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17e5caceb132f6da6c283dbdd5fcb201db04bde4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 06:29:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/026f55c06dbc4e7b84efb5c10d02d78d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/954e0cdd79d848a3b290311bdce8636d : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fd84562a39034c1c9162a404634c593a : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8a5577a6e22943749faa5a81d3783ed5 : SUCCESS in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3683787ef2b9ea1671d059dd2c1d33472f8abae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 06:35:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08f0d6c0b491424eaf3411dbc471ef4d\n\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ac750471980d483592c7f75adf0c797a : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/554d15636618460196f4860ee21348b4 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ef4ba87500b46dfbe28aa1e67b237bd : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab2e56b15bcf46c6b056d588d4705770 : SUCCESS in 2m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/938c29c7b9884c40ad2b6700e413db3f : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35b6996114464297af510816bb40c02f : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38397380707c438595e6ae083090bd16 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1639b3becd524077bf1a9b6d0bebb902 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0f71f50369b7b62800d81652531514c895ac6f0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-22 06:47:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"39016e8f119e8891000415d89b5e95ca441fd56b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 06:47:45.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  The secret \"kolla_copr_erlang27_jul_2026\" was not found.\n\nThe problem appears in the \"kolla-copr-erlang-update\" job stanza:\n\n  job:\n      name: kolla-copr-erlang-update\n      parent: base\n      nodeset: kolla-ubuntu-noble-8GB\n      timeout: 3600\n      run: tests/playbooks/copr-erlang-update.yml\n      secrets:\n        - name: kolla_copr_erlang26\n          secret: kolla_copr_erlang26_jul_2026\n  ...\n\n  in \"openstack/kolla/zuul.d/periodics/copr.yaml@master\", line 2","accounts_in_message":[],"_revision_number":2},{"id":"54d281df5f77fa2ae5ca11f9b06e509cec1216e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 06:48:00.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  The secret \"kolla_copr_erlang27_jul_2026\" was not found.\n\nThe problem appears in the \"kolla-copr-erlang-update\" job stanza:\n\n  job:\n      name: kolla-copr-erlang-update\n      parent: base\n      nodeset: kolla-ubuntu-noble-8GB\n      timeout: 3600\n      run: tests/playbooks/copr-erlang-update.yml\n      secrets:\n        - name: kolla_copr_erlang26\n          secret: kolla_copr_erlang26_jul_2026\n  ...\n\n  in \"openstack/kolla/zuul.d/periodics/copr.yaml@master\", line 2","accounts_in_message":[],"_revision_number":2},{"id":"3aa48d5ba1c7ec4a8020093471e9896aa7495e6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-22 06:52:47.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"bea2bc69a353efba76da7ee68ad0132b55a56583","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 07:01:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b130bd6b02b540a49ead4ddab09c6a58\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/90dfac307c834dcda137c6d8a8b4c753 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/efa9f4a2d6a74c38b1d5cb66d2f07c06 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/df4005321da8447695e2e6a27957a97e : SUCCESS in 6m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1167a355d3f3511bb65c9bc4091f6a05f920d644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 07:04:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b67a21a1fe64dd8bc1a2923bb238b9e\n\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bfabd1e7ef574bf5bbbb6237b3e73a80 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49ec2ebe7ac147639b91efcdc52d58fd : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6231088d11bf4c6c86fc3e016d353f0b : SUCCESS in 5m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c8b3552a55a4ffb8ce7e674c2eb64bf : SUCCESS in 2m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3ace428af72d47aab141e8a333a0b291 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df98e1aa921d4843a6e14cc2360259a7 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c30829ed89b4d038664b4bb97ad6b5d : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4e4e88082ab443aa3490e9816772111 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"d18428339e70fd50f8031b7f1d5a42f11fa29d28","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-22 07:09:38.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"233c6c672156afe4f8f14287a4cee4355f1ee95a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 07:10:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e9a11d491e0f763a156964d686d991dc7c2e877","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 07:24:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92453ed572ce4ae9a6b1add6ddd9d2ac\n\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/35e209123dbc4a1492c4fa8475dfeb5b : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8acb92472c89441fac2ddd1814127c58 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/976f9d37ecaa4774a6de2745eab3a658 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef92ea2ead354caf8cb0a46d6281640c : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29b2eebb770647c49d1815cc2b1dad6b : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfd8a13ead114ea694c9ca6862b1c2f8 : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":3},{"id":"22b584369e4865be162f85ad1f7419ab395911f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 07:24:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"45a11cc7fbc2615060a62d64fec274635c193071","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 07:26:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea8a5c360fef42d3a556dd9e1b5cde36\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79df76b5465947f2ada3e2f50c729c98 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"144c51bcac340f3f1c1f8134ba6af722b85a7b7e","revisions":{"599dfa455e167c9b10f831b6f68bf5a3c53a1d68":{"kind":"REWORK","_number":1,"created":"2026-07-22 06:22:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/998266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/998266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/998266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/998266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/998266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/998266/1"}}},"commit":{"parents":[{"commit":"258b35e3a3ff609bd3f6344c98cea3912ab55cfb","subject":"Merge \"rabbitmq: Update to 4.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/258b35e3a3ff609bd3f6344c98cea3912ab55cfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-22 06:19:00.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-22 06:22:31.000000000","tz":120},"subject":"CI: copr - stringify part 2","message":"CI: copr - stringify part 2\n\nChange-Id: I37bb0b5056b2e055d0d0c9da5d0beba4b9ba9067\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/599dfa455e167c9b10f831b6f68bf5a3c53a1d68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/599dfa455e167c9b10f831b6f68bf5a3c53a1d68"}]},"branch":"refs/heads/master"},"ce7e3a559c98f2e18891703ce70c84db5d9b288b":{"kind":"REWORK","_number":2,"created":"2026-07-22 06:47:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/998266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/998266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/998266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/998266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/998266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/998266/2"}}},"commit":{"parents":[{"commit":"258b35e3a3ff609bd3f6344c98cea3912ab55cfb","subject":"Merge \"rabbitmq: Update to 4.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/258b35e3a3ff609bd3f6344c98cea3912ab55cfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-22 06:19:00.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-22 06:47:10.000000000","tz":120},"subject":"CI: copr - replace secrets without newlines","message":"CI: copr - replace secrets without newlines\n\nChange-Id: I37bb0b5056b2e055d0d0c9da5d0beba4b9ba9067\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce7e3a559c98f2e18891703ce70c84db5d9b288b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce7e3a559c98f2e18891703ce70c84db5d9b288b"}]},"branch":"refs/heads/master"},"144c51bcac340f3f1c1f8134ba6af722b85a7b7e":{"kind":"REWORK","_number":3,"created":"2026-07-22 06:52:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/998266/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/998266/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/998266/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/998266/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/998266/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/998266/3"}}},"commit":{"parents":[{"commit":"258b35e3a3ff609bd3f6344c98cea3912ab55cfb","subject":"Merge \"rabbitmq: Update to 4.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/258b35e3a3ff609bd3f6344c98cea3912ab55cfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-22 06:19:00.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-22 06:52:46.000000000","tz":0},"subject":"CI: copr - replace secrets without newlines","message":"CI: copr - replace secrets without newlines\n\nChange-Id: I37bb0b5056b2e055d0d0c9da5d0beba4b9ba9067\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/144c51bcac340f3f1c1f8134ba6af722b85a7b7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/144c51bcac340f3f1c1f8134ba6af722b85a7b7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
