)]}'
{"id":"openstack%2Fkolla~998606","triplet_id":"openstack%2Fkolla~master~I3388fff341f498ef809d02aa2b4badae0da7ca17","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I3388fff341f498ef809d02aa2b4badae0da7ca17","subject":"ironic: Build a multi-architecture ESP image","status":"NEW","created":"2026-07-24 11:32:10.000000000","updated":"2026-07-24 14:10:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"cb58adf01f3a8cb51b113c0280862a9ae7695545","_number":998606,"virtual_id_number":998606,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-24 14:09:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-07-24 14:09:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c946f5f9b12ea6257b818f5bc54d0717e9a81ccd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-24 11:32:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db3e75450c464ef1259cdbb9584cdb9464be1662","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-24 11:32:18.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ede9beb4fbf90f5eab6795473b4d19fcdfb58073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 14:09:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/503c82b1ee8b4d2d9bc46d6b7530a7db\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/bc918d6e511347a49476f87c2637ae09 : SUCCESS in 54m 12s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/c4e33f1f3e39490a98840e68dd6ce28b : SUCCESS in 23m 06s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1b6aca6b1721473dbc83f2745ec71922 : SUCCESS in 25m 14s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7ab968f14bca4bbba97424e911b684b9 : SUCCESS in 16m 56s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e2f39c1a243749a7bd1442c4b4c778a3 : SUCCESS in 52m 50s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d19c000c99944ae08cc0ba9128009dae : SUCCESS in 16m 22s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8400aa92513b4dc68f9514944d7c7f91 : SUCCESS in 27m 07s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/03ef89740d084f80ad63754e43bdb81d : TIMED_OUT in 2h 34m 19s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/6c028bb479cf46b69e1cbaf61a36c866 : SUCCESS in 1h 03m 26s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/1b6e8586b2944e478dbb4a4b2e6d8dd4 : SUCCESS in 1h 27m 57s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8d01d747edea4ab29cbfb867b968971b : SUCCESS in 1h 04m 37s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/dff3622802bb4d60a40948af52bbcb31 : SUCCESS in 1h 25m 17s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8514721c3914d689e67716b4d3a3ebf : SUCCESS in 1h 04m 13s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/5a2fc699836044fa9874bdc1d3fca25c : SUCCESS in 1h 37m 09s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/a21af786623442b090e32a6081d55fc2 : SUCCESS in 1h 21m 05s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/ce5f83edf2834208b92e996a8f7dc25f : SUCCESS in 1h 18m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/5f9a097c195445e59319c2e096d1fecf : SUCCESS in 1h 19m 58s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a8d2245f50bc49a49eacda8266db9b4c : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c99fc124a34e94a7e96df6586224b2 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32ca0399c6a44683bd2ff4b253dfc9da : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fc9f9ee772d4d1aad63be97b82d83d4 : SUCCESS in 5m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b09b8324d8db49a5bd4c89cb69e4d346 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad4a7499b5184ba085d7607a7a8f466c : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a1aee64c2794d70ba92509963a39d3a : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f181d0dfbf44d9ca74c07d8d5fef78c : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"cb58adf01f3a8cb51b113c0280862a9ae7695545","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 14:10:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/302630747be54a74a8880a1fac8dfc0d\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/592bd8f5e2704a8dbd6a491b60fb2125 : SUCCESS in 40m 09s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/db11a8237d5d45f390c4c669244cce3e : SUCCESS in 39m 35s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b735279f6db843a5bf75cb0687b32032 : SUCCESS in 42m 30s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/04cc2fa85d044ecc955d6749640e2dc6 : SUCCESS in 35m 03s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/6f72c02d84af4b6e91b05dfe632d3158 : FAILURE in 1h 27m 01s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/f0d3a5e27b7a4574a01661c4cbe15708 : FAILURE in 1h 16m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/30fc07c6118c4b4fb06a5183e4987edf : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/953cb156446547ccb9e59fc59cdac680 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/897d7c9d841c48ccae140dcf6511d061 : SUCCESS in 8m 56s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"941d9b1f0895b624d3e55a9ce5da73c0c70519d8","revisions":{"941d9b1f0895b624d3e55a9ce5da73c0c70519d8":{"kind":"REWORK","_number":1,"created":"2026-07-24 11:32:10.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/06/998606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/998606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/998606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/998606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/998606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/998606/1"}}},"commit":{"parents":[{"commit":"5d77668da8167435e4ac860392994f19ec733a46","subject":"Merge \"Rework tar normalization to a shared function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d77668da8167435e4ac860392994f19ec733a46"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-07-24 11:32:01.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-07-24 11:32:01.000000000","tz":120},"subject":"ironic: Build a multi-architecture ESP image","message":"ironic: Build a multi-architecture ESP image\n\nInclude x86_64 and AArch64 bootloaders in the ESP used for virtual\nmedia boot. Distribution repositories only expose native architecture\npackages, so download and extract both variants during the image build.\n\nAt bootstrap, extend start only copies the prebuilt ESP to the httpboot\nvolume.\n\nChange-Id: I3388fff341f498ef809d02aa2b4badae0da7ca17\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/941d9b1f0895b624d3e55a9ce5da73c0c70519d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/941d9b1f0895b624d3e55a9ce5da73c0c70519d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
