)]}'
{"id":"openstack%2Fkolla~999090","triplet_id":"openstack%2Fkolla~master~I2d8f88cc9a4c022f531587ff920fde579564cd80","project":"openstack/kolla","branch":"master","attention_set":{"37203":{"account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"last_update":"2026-08-05 13:02:06.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-08-05 13:02:28.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-08-05 13:02:06.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2026-08-05 13:02:06.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"I2d8f88cc9a4c022f531587ff920fde579564cd80","subject":"Add step-ca container image","status":"NEW","created":"2026-07-29 11:58:00.000000000","updated":"2026-08-05 13:02:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9ba4183c350f7431f361f296a6be9c70bdcbcbfa","_number":999090,"virtual_id_number":999090,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-05 01:38:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-29 17:28:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 13:02:06.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-08-05 13:02:06.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"state":"REVIEWER"},{"updated":"2026-08-05 13:02:28.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"0147b18a0c5f00844d51f303b562ef22aaef04d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-07-29 11:58:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc87b44ca13df39dcdd338989c9271fae968a549","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-07-29 11:58:12.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"804674a9f6245da423ee3964e31322169186b7a1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-07-29 14:47:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"034c42f3c5ced8edfb015323d60a715b879e74a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 17:28:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0be34aa141e24db0ab5f38fb762fc673\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/712fc2f22824418bae254d48eeb27b1a : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/c94756bc348d43518bdb0de2f55a204e : SUCCESS in 26m 30s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/84f551fae3cc4788846d456b3ca785c8 : SUCCESS in 13m 53s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bd7ac77ec17149c7a3c4ad0008308d84 : SUCCESS in 15m 33s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e17ab3f14af14721b8ca9f5020aafeb0 : SUCCESS in 57m 05s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2e9fd476b97466cb193df390017b3d5 : SUCCESS in 27m 27s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c2c3c4fb6eb04e8c9b65997d42cc1d93 : SUCCESS in 30m 06s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/9220ea89b9b44aa9b2161759939fbeba : TIMED_OUT in 2h 34m 55s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/eb2924e619034c10aea88f6ec10e8911 : SUCCESS in 1h 04m 58s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/4c985fb21770485dbbe9a935b2ef8251 : SUCCESS in 1h 27m 32s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8357aedf684b41d59d92088155f7dcc3 : SUCCESS in 1h 07m 12s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/4946ccff385f4648b9887cb045460ba4 : SUCCESS in 1h 41m 13s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f8584fd6b264baabbd1efd961d4598d : SUCCESS in 1h 08m 19s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6dd8affae9ad46bfbc8af0b8b9b96c3b : SUCCESS in 1h 33m 18s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/a58b85d08f704beda022ee45e57777bd : SUCCESS in 1h 36m 16s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/5df35370cf0c4108bae2b27e9ecda749 : SUCCESS in 1h 42m 27s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/625f3af39ee14dd288a8939d27395f44 : SUCCESS in 2h 19m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/70013b28af494a98af12d327a170e47a : SUCCESS in 1h 45m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/eb1f88aef79640adb73dbeb81d97d496 : SUCCESS in 2h 28m 00s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/3b29bb1b91cf40f5b942aeb2e4070956 : SUCCESS in 1h 18m 49s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/b846f76963fe45679dc1c89803288bc5 : SUCCESS in 1h 29m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/4b1174a576d54267be1f67f5ee4881f8 : SUCCESS in 1h 33m 05s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/961bc22475d94f1682fbe60f42fe17f3 : SUCCESS in 27m 11s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/245a3ac2363245f595f581c474333c30 : SUCCESS in 26m 03s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/f6949f2e4f934f4db4ad00330bbde819 : SUCCESS in 35m 09s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/070f9315c8354e3faa483d12b8366dc5 : FAILURE in 1h 02m 44s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/4bd23969b22046e8a2072df9f9397512 : SUCCESS in 44m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/8c21324840204475a1ce22e87aba561b : SUCCESS in 39m 17s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/575dc9d466684490a41b66e01ac4a6ca : SUCCESS in 1h 16m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/58db54fd14a84a0a8639e603305dbfde : SUCCESS in 1h 07m 37s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/549987174cc249cfa5c17c408cfdabf6 : SUCCESS in 1h 39m 06s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/f3a3f872fcea47f2850f7d68ccce3897 : SUCCESS in 1h 36m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/5b19dc4d0c474f5ba311d2a6f303fcfc : SUCCESS in 1h 49m 16s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/3c0d4cbbb4d2488180ea865f0b8a4590 : FAILURE in 53m 22s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/9412b4c785c149fc8d2daa9e9c1ac727 : FAILURE in 1h 03m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/f3a60b1fb4404c9dbb69f8e958785373 : TIMED_OUT in 2h 32m 04s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3ed8d1538aa54598930b1d9fe1d19e59 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47c897d885674340924ab2ec14cf466a : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/638fd404e7d84ebeae6c6263f7839777 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69890d775a1e4adfad4382b2da0108a0 : SUCCESS in 2m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/59b8a311eb6b4aa1b284315bcd6b4274 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4b7fab199324a69bb2ecdc15b9abdb1 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ce05f74f424c4e82c209e7ea45c81a : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/478da04b28bd465db29cf8530ca2c3bb : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"41e51230f886e1e37567e9889009245d9da9a190","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 07:27:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50972fd925a94e229ae42ba406017487\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/769ef106b2c54db2affe198fdfaeaf4e : SUCCESS in 1h 51m 39s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/1693a67e10644205b526708c8ef1a813 : SUCCESS in 1h 55m 11s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6d30248b46a846e292d5d71709142ea5 : SUCCESS in 40m 30s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4c5298f0576947969b602ace5ebaf57a : SUCCESS in 2h 02m 28s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/d2c89aa410994b5792c28a4adcc48301 : FAILURE in 1h 24m 15s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/4a0b9dca2b7043b1a4422d030f5a41f4 : FAILURE in 1h 33m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a5c0aec3e02f4a0d96d27c3ba9bd613c : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b364f560a1e049d682a37bfdd8d56e09 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/95cf56dbc9624b39a4bde6dfdfcf7f3a : SUCCESS in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"98f922a09c08a7852898f8cb1275ee1b685161b6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-08-04 08:35:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"aff3ca11ad4e9c5d0151d1864daca11640f175c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 09:52:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b87748291c34651bc04be299c9ab08b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/6fd7dab369404a3894f585684f5caa37 : SUCCESS in 39m 19s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/b3a5a1b6cb3c4db2912b40206d06a400 : SUCCESS in 34m 57s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/21463577264944de9563225b5ce55fbf : SUCCESS in 36m 10s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/0ac948c63d624cff89316ed45bb8fad6 : SUCCESS in 39m 12s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/41f1e4202c3a40f9bcfbb82d5773b5ec : FAILURE in 1h 15m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/33ea3e76e0b04f15a8ea648c36c05fd0 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/88053969fd524a6e9e5166b15ff87b7b : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/93217ac168c74092beb093377be01de5 : SUCCESS in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cecbdf0f574f7439d38c9fe88be5938b09d42582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 10:59:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0255ab2b754c4d2cbd9fa94260c3c158\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/bc269957079d47c994dddb420d5f1428 : SUCCESS in 55m 44s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/4bdac84a8d474c65a6c9afe8af5e23d5 : SUCCESS in 21m 48s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ea50623756824f40a429d1cf79442b04 : SUCCESS in 20m 16s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f5b794cc73724cf9ab02d6b931ec051e : SUCCESS in 16m 29s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/1794250ae034493abfd06b485b2cd8f7 : SUCCESS in 28m 25s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3504475c1fc4ce6aa78dfe64496b69d : SUCCESS in 19m 39s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c70b40064dd441d191772236ac79b37b : SUCCESS in 13m 55s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/b5f7e4cb6dd44f508f50ccb681e944ac : FAILURE in 42m 54s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/2a3e68d1f2d040b88466664b332905f0 : FAILURE in 1h 34m 07s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a78b31571e534b7f852e9db8f0289f05 : FAILURE in 1h 01m 05s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b7368204c5cf4848890b638f90adae35 : FAILURE in 1h 50m 03s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcfba2160198481b8d36647cdbeb858f : FAILURE in 46m 03s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/a5a5dc828a4448d0815340dc22766d24 : FAILURE in 1h 50m 08s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/27f0bb60c0f94238bea68eccb18f8d3b : FAILURE in 1h 10m 03s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/85b40dfd642d415b98fc3315854c28a5 : FAILURE in 1h 12m 22s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/03dbf1e4e19c4e43999e9525be8f9144 : FAILURE in 2h 21m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/45d098ef14204c2b98e6f168a9438f7d : FAILURE in 1h 22m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d883647fa88f42009a07929b54e5d8ee : FAILURE in 10m 57s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/e5740dfff6254f87bbfe207da9876814 : SUCCESS in 1h 18m 36s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/9c74d6df049e43bf9992f8be8e1e0370 : SUCCESS in 1h 21m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/d64370fb992343499d2128866331f3e0 : FAILURE in 1h 04m 17s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/42941870285e47c28d7794bd7fb9c7fe : SUCCESS in 23m 41s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/f9f98c48d755488aaf6100f3745c0b54 : SUCCESS in 23m 43s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/14a26aa438094f8b8c59d252780a4455 : SUCCESS in 30m 33s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/b26d650eb7fb4a489d75ebfcc31e4f6f : FAILURE in 9m 42s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/402ddc85b9604fac8a92884381ff8ba5 : SUCCESS in 38m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/cd031e9e15d74f92a31728dfdb5af870 : SUCCESS in 1h 18m 20s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/0243b6c7b80049caac851c04b60a172c : FAILURE in 39m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/af70a8cedf7b4f428276a6b35332fe24 : FAILURE in 42m 44s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/7bc67e8ace8c49a2a7f39b6b29b6d571 : SUCCESS in 1h 33m 57s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/4e4ecbddc91d4e0299f6e5361ba0eee8 : SUCCESS in 1h 39m 02s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/f4a59e0f928b4e3899b00d583e5992dc : SUCCESS in 1h 44m 12s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/22e0a176821f4010b9b6d9c0b41eca23 : SUCCESS in 19m 19s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/cd859920ba604b5182146a94a4bcee35 : SUCCESS in 55m 49s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dab376803e8d4f36aff6f5ef0e631767 : SUCCESS in 34m 25s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/636e0b6ab97a41558308389f61565986 : SUCCESS in 48m 47s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dfc838b588384aeba7e9f19385d2ab42 : SUCCESS in 39m 45s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/7574dc3fdb614a5f8ebb6a4419d16a81 : SUCCESS in 45m 18s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/676963eefe654bd6a38cf7e529db1abb : FAILURE in 1h 08m 51s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/7d4d735ff2ce46318f8a130b8c84c28d : FAILURE in 53m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/1e8608dca3e44bff90acee2d489668f3 : SUCCESS in 1h 30m 19s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1f8d615669944cd39f5fddb39da26ca0 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08165648f2ae4063a5242a05f4560d02 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf915fe86c554f2aa454e0e5e82afc17 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bdd9e8a0febf49f28d2af0b3fe34437b : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/681216a52bf74559a88b876d234d06b6 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7abf14257b74c96aaf55b9464524a39 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33243b03f2c14d8793a1c329dabe07c4 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/731cbafe63f94172bd3e8fb48c0bf7f4 : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":3},{"id":"6d12404e2562bf603f29abca8a559c9854d9cefe","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-08-04 16:34:11.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"4d16d57f2ba6de454da755ddb6f179b7656e02ca","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-08-04 16:34:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ebd86df2d5160aa2000cf30532995886ff8236ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 17:35:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03a2a9b517784b4cbd59304599a112a3\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c0ccedfe2125425f83d88e93839c055c : SUCCESS in 42m 45s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d608b3846a1d4b73b961f972a28baef0 : SUCCESS in 31m 32s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b22e5a957e1840be90f7eeb03b4c0056 : SUCCESS in 37m 27s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/1b026ec80289459abf6119f6d9a78fc2 : SUCCESS in 27m 59s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/b7dbb4e7065646859f89eef13eea2e2a : FAILURE in 1h 00m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7ddf5fb51e8943a7961e94fea37178db : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c388a0a864c7406d895c18937be122c3 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/da7c9d5f7c7b4743a926b304f31ae03c : SUCCESS in 7m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"431cf5e0578a0cc08eaeb3213acab18299cb933c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 19:29:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7760e863013c4a7c9f750815515642da\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/cb93517e413f477f99aab2aaee538296 : SUCCESS in 56m 39s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5bd648d05cf94f58b51544ddaa12ceaa : SUCCESS in 23m 57s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9700566fade243e0bbc2c20d96d904c7 : SUCCESS in 26m 57s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d86751771cb3472ba671522d5ebc2b34 : SUCCESS in 20m 40s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7104146a859f4a65a69e5f72ee542283 : SUCCESS in 50m 46s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/562f7bffd96242c3869a51a286502199 : SUCCESS in 32m 19s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3f9827b81648430091d13bd4acac7a55 : SUCCESS in 27m 51s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f07a10fa7cd4407984b8165ddbc67e70 : SUCCESS in 1h 18m 26s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/e44122f7ca074e87ab5875ce967e14c3 : FAILURE in 24m 57s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8d21d303b0ba4ab8a130e7a021922841 : SUCCESS in 1h 19m 35s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/473f51bf6f464326898e8d8eab5a4697 : SUCCESS in 1h 47m 54s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa68f1b04b484664b9c0b32fd7efb71c : SUCCESS in 1h 22m 07s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ed40d2fca6444e948e87a45c905174d9 : SUCCESS in 2h 11m 43s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/46015f9210f445b0a14402a84e91392e : SUCCESS in 2h 03m 08s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e99accaa926c45448409014531ab19f3 : SUCCESS in 2h 07m 58s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ee21e909c1914dc9b0b7ff02bea85f6c : SUCCESS in 2h 52m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/562a481f9b4241b59587a9c44ec2c4f5 : SUCCESS in 1h 37m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/a9e4367dd1804a8499c41a24be06eaa5 : SUCCESS in 2h 13m 07s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/d0fd1af0eab94ee89395111a0e0210dc : SUCCESS in 1h 21m 45s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/0373cb01c2e44ef1bbb4fbd47c5254a6 : SUCCESS in 1h 23m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/809350f5636442b8b634e18404abb934 : SUCCESS in 1h 31m 57s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/336887621ae4475fb04750abc69f0f08 : SUCCESS in 24m 29s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/73587bc6a9644231b9571d111117ae46 : SUCCESS in 30m 54s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/7bc28337faad4fd18ef7e4116f2b79b4 : SUCCESS in 33m 49s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/6475a0b356c24ea6bbf8211acdc1b827 : FAILURE in 52m 44s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/7f8ea73874d0452c9895f320ecfc9021 : SUCCESS in 1h 03m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/0162d7fd2e8d47b785512ade08f8472f : SUCCESS in 1h 09m 31s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/3454a5e6824645528ffe174c7218c1b2 : SUCCESS in 1h 08m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/ce0381f30d43478ab35f207cbde04287 : SUCCESS in 1h 14m 07s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/a968bcb2cb0a4c63adf0558c9b0de09e : SUCCESS in 1h 45m 23s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/096163d040594c63994b7e352dd2352c : SUCCESS in 1h 36m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/d730d6cfd2e24fe985cc0c2dcee40693 : SUCCESS in 1h 41m 01s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/efc3ab73f7c6435b955fc4e2ae44e37f : SUCCESS in 24m 43s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/5d2db6fc50734a5697e5ca77135c5abf : SUCCESS in 52m 07s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/951b8b5019cd431c9c09b96d5eef1ed8 : SUCCESS in 42m 32s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/38cc58da909145239b055bbd3eaaafbe : SUCCESS in 44m 57s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/295aef8200c54914a4428ae5515c4d0b : SUCCESS in 30m 37s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/dd5106ecfb0341a1b9551b270dff1248 : TIMED_OUT in 2h 31m 22s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/1c91ad82025349389040b66d331703c3 : FAILURE in 1h 18m 35s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/6da6ca65dfbf47f9b0110d91e5da99cc : SUCCESS in 1h 13m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/e944671990be4530bfb7fb830f5c6ecb : SUCCESS in 1h 37m 35s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/07455b4c8d7d4d24a745a04797ca603b : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f0117c3cc94cf0b5f92f7efe4bbbfe : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/409e517d6054481389969ef01a36af32 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be698d509c514c16ad287a82fcff5be4 : SUCCESS in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/14109f1dc429492589a4e270609b3f4e : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b37dd340d1974f6e956bafac8356d5cb : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd1d9a8ebf684e709601daf61f650aef : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5717e7cfe584d55bcb811b066d5bcbe : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":3},{"id":"314699513ea52ee2eb866110d42e105ed8da2a0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-08-04 20:38:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"b46ee375aaee407edb60f040b86e7a9fbf025761","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-08-04 20:48:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ba22b4799abe585473ceaa937ba7fb1c6cc12b65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 21:56:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f68c50300b44fc7b41dfb9da2a57911\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/f3f4df712b2143cab23b4e3b8204ce53 : SUCCESS in 57m 45s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d589efda55884ae3a8ebb4a9dfd34f64 : SUCCESS in 55m 21s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6b0dc7ebde7740858411c62507459e2a : SUCCESS in 45m 23s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4cff1c7e016c4cf698a0c9dd22bcfbbf : SUCCESS in 1h 00m 03s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/784aec9ed3cb4a268f95d8bb93f36724 : FAILURE in 54m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7af06ccea9314f3e90d7d1446f559c13 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/40726e7251854f0c8ae8a744e4a589f6 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/30be9a35ba194066854ac58b3cf69ce2 : SUCCESS in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5e49ae24fddf54c0552b109d04e47228e4d47726","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-08-04 23:05:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8d867bdedfa35cdd0df4787176d662c600bcf4cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 00:26:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3712a8aa772b4637ae9835751cb94898\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/e659bd0b40394cefba10d39abae5fe0a : SUCCESS in 39m 50s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/0adc96425e41484c81d04f687a475d4e : SUCCESS in 23m 55s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/dbf0ff92161d46f8b9da68d5782c335f : SUCCESS in 33m 30s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/68ee8155431f48cf98543b8a3cb6fd8d : SUCCESS in 35m 53s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/0ea085c2881e470a9a966b825114530b : FAILURE in 1h 18m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/55e5b26108f94b0ca76c394bc5672424 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8d79e138ae4947e2adaacc1db76a7a31 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/382a5234580e4d37b4487e1fae014f9d : SUCCESS in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f3c6d5624faab7e1d2eb908a8dc3773ffecbac6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 01:38:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/187a868370e34e4c8883b42e13741bd9\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f53ae89550d84af49dced75bb94e9d64 : SUCCESS in 53m 22s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/907ae5dbae794600ac69e9c43f864f79 : SUCCESS in 15m 54s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/49f1249e1cc54b7fbd026bc5d5e132cb : SUCCESS in 30m 46s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/19f03c9f098e4111ba6558bc7b5e2041 : SUCCESS in 14m 04s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d07657b5f48a4726a4f09f26b0e99d03 : SUCCESS in 41m 49s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c53547554eed4b52bef267d30d895455 : SUCCESS in 17m 36s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/944b9f11af7e435e824d21c149d55b48 : SUCCESS in 18m 14s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f2fbf3abc3264a64b3087aee2453c8d0 : SUCCESS in 1h 01m 54s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/863ebd3fdcda4284b276bf33dc6a9976 : SUCCESS in 1h 32m 19s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a314d59e217a4a659b5dcb40af062c91 : SUCCESS in 1h 02m 26s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9c3b5f1fe732476197696adf5093c902 : SUCCESS in 1h 29m 16s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff3819078a4f454881c54bbc71c3cbb7 : SUCCESS in 1h 14m 20s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/73425476bfcb40698b083a239ae517da : SUCCESS in 1h 32m 56s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/891c8aa9b3264a0db096c8fdbf823228 : SUCCESS in 1h 26m 47s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/db2125fd4147495f9232788f85d4aeed : SUCCESS in 1h 50m 27s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/e91c1d41a91741cbaeeb74335e0ffae7 : SUCCESS in 2h 30m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/e596d9b895b147f884d3d14ff1223fe2 : FAILURE in 12m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/3f6a8435b6d741f882c3a0d5c7d0e511 : FAILURE in 2h 03m 23s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/c0d50bbd64ec438ea0bd6ad6b490a2e3 : SUCCESS in 1h 18m 16s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/9493e1905ae74d43964c8f7b890b560f : SUCCESS in 1h 17m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/3f9567e07a4c4852ae964ba2a7aa5039 : SUCCESS in 1h 23m 59s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/bd515069782b4dde871833dd17734448 : SUCCESS in 24m 06s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/7dba3fbb9d174527850b6037a864c6eb : SUCCESS in 28m 13s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/a311d6feccbd4a9cabac9d0638c55353 : SUCCESS in 30m 40s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/c2ba6d05b76b4a498b9b568a16814b26 : FAILURE in 27m 24s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/a90e450b05b34644a89b0afc749be773 : SUCCESS in 59m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/c03a637a19024d29ab75268801b9d1cf : SUCCESS in 1h 03m 33s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/1167716f699e42e69703037c4272b9f1 : SUCCESS in 1h 07m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/861e2ff0aad34b608755238db9f9db71 : SUCCESS in 1h 07m 46s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/2eaf3c32e926413abba144eb231ada5d : SUCCESS in 1h 40m 27s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/b90d4a7464d34546a06713174f081bf4 : SUCCESS in 1h 39m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/233f3feb73cb4e4292243dc4a44cf1e9 : SUCCESS in 1h 41m 18s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ecfce0f6acac47c99dcf57a13ef34475 : SUCCESS in 28m 23s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/d503b4eb8bfa4b5f953387b3a12c1c69 : SUCCESS in 53m 56s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c09ec200fea44e8a9817203d52ab24c1 : SUCCESS in 32m 28s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/d5d47e82b5e44b1e880657eea3bbfec0 : SUCCESS in 47m 32s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dc27ec53efc14ad4bddd2a0352acd546 : SUCCESS in 38m 55s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/f7ba7a498df6441297ef7927d50eda81 : SUCCESS in 55m 04s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/890ef513b591408fa94f1da882f1487e : FAILURE in 1h 12m 28s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/9fef95277e104eaea1da4fdc7d0e00aa : SUCCESS in 45m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/9982c0e7d07645f693003a72ecf4de39 : SUCCESS in 1h 29m 11s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d227499286e1455a845ad303516539ab : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/803538946de246ebaf25c0c6fe768119 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a9642b264e8493c9aa7f11eeb21ce5c : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed8d049e29134319b526b5058c516ed6 : SUCCESS in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/70cd09e3311c4131a65d5262c941c1a6 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f885e738bdaa4635b651d7b235e08b05 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b16e6f3945ed408ebe1338d25ca506f9 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/904dcf3ae9bd4f1db0072ecd62d50bb2 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d3aa6b49c5abd5f2c914efe26efeb589576514e4","revisions":{"390363ebb302206f11d0ed31c18ffdeaa8ea700a":{"kind":"REWORK","_number":1,"created":"2026-07-29 11:58:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/90/999090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/999090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/999090/1"}}},"commit":{"parents":[{"commit":"5d77668da8167435e4ac860392994f19ec733a46","subject":"Merge \"Rework tar normalization to a shared function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d77668da8167435e4ac860392994f19ec733a46"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-07-29 11:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-07-29 11:57:28.000000000","tz":120},"subject":"Add smallstep-ca image","message":"Add smallstep-ca image\n\nChange-Id: I2d8f88cc9a4c022f531587ff920fde579564cd80\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/390363ebb302206f11d0ed31c18ffdeaa8ea700a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/390363ebb302206f11d0ed31c18ffdeaa8ea700a"}]},"branch":"refs/heads/master"},"c9c2169f629a8bd7b7cabe08aba065d37f8cc614":{"kind":"REWORK","_number":2,"created":"2026-07-29 14:47:58.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/90/999090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/999090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/999090/2"}}},"commit":{"parents":[{"commit":"5d77668da8167435e4ac860392994f19ec733a46","subject":"Merge \"Rework tar normalization to a shared function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d77668da8167435e4ac860392994f19ec733a46"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-07-29 11:57:28.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-07-29 14:47:52.000000000","tz":120},"subject":"Add smallstep-ca image","message":"Add smallstep-ca image\n\nChange-Id: I2d8f88cc9a4c022f531587ff920fde579564cd80\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9c2169f629a8bd7b7cabe08aba065d37f8cc614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9c2169f629a8bd7b7cabe08aba065d37f8cc614"}]},"branch":"refs/heads/master"},"a941812d726d4e9cecc22f8ad70b90a4e91a4cef":{"kind":"REWORK","_number":3,"created":"2026-08-04 08:35:36.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/90/999090/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/999090/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/999090/3"}}},"commit":{"parents":[{"commit":"8e912a9db7f013a1abfc1e70a829e435d50622db","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e912a9db7f013a1abfc1e70a829e435d50622db"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-08-04 08:34:39.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-08-04 08:35:27.000000000","tz":120},"subject":"Add step-ca image","message":"Add step-ca image\n\nChange-Id: I2d8f88cc9a4c022f531587ff920fde579564cd80\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a941812d726d4e9cecc22f8ad70b90a4e91a4cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a941812d726d4e9cecc22f8ad70b90a4e91a4cef"}]},"branch":"refs/heads/master"},"3fdcca97855890d1c982f4cbaf849e3fb97eb1e3":{"kind":"REWORK","_number":4,"created":"2026-08-04 20:38:22.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/90/999090/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/999090/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/999090/4"}}},"commit":{"parents":[{"commit":"8e912a9db7f013a1abfc1e70a829e435d50622db","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e912a9db7f013a1abfc1e70a829e435d50622db"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-08-04 08:34:39.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-08-04 20:37:41.000000000","tz":120},"subject":"Add step-ca container image","message":"Add step-ca container image\n\nIntroduce a new step-ca container image based on the Smallstep\npackages. The image includes step-ca and a wrapper\nscript for file logging.\n\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nChange-Id: I2d8f88cc9a4c022f531587ff920fde579564cd80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3fdcca97855890d1c982f4cbaf849e3fb97eb1e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3fdcca97855890d1c982f4cbaf849e3fb97eb1e3"}]},"branch":"refs/heads/master"},"da3555398fa37bc8c6b9371e84a0e163b70719ea":{"kind":"REWORK","_number":5,"created":"2026-08-04 20:48:40.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/90/999090/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/999090/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/999090/5"}}},"commit":{"parents":[{"commit":"8e912a9db7f013a1abfc1e70a829e435d50622db","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e912a9db7f013a1abfc1e70a829e435d50622db"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-08-04 08:34:39.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-08-04 20:48:34.000000000","tz":120},"subject":"Add step-ca container image","message":"Add step-ca container image\n\nIntroduce a new step-ca container image based on the Smallstep\npackages. The image includes step-ca and a wrapper\nscript for file logging.\n\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nChange-Id: I2d8f88cc9a4c022f531587ff920fde579564cd80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da3555398fa37bc8c6b9371e84a0e163b70719ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da3555398fa37bc8c6b9371e84a0e163b70719ea"}]},"branch":"refs/heads/master"},"d3aa6b49c5abd5f2c914efe26efeb589576514e4":{"kind":"REWORK","_number":6,"created":"2026-08-04 23:05:37.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/90/999090/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/999090/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/999090/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/999090/6"}}},"commit":{"parents":[{"commit":"8e912a9db7f013a1abfc1e70a829e435d50622db","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e912a9db7f013a1abfc1e70a829e435d50622db"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-08-04 08:34:39.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-08-04 23:05:29.000000000","tz":120},"subject":"Add step-ca container image","message":"Add step-ca container image\n\nIntroduce a new step-ca container image based on the Smallstep\npackages. The image includes step-ca and a wrapper\nscript for file logging.\n\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nChange-Id: I2d8f88cc9a4c022f531587ff920fde579564cd80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3aa6b49c5abd5f2c914efe26efeb589576514e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3aa6b49c5abd5f2c914efe26efeb589576514e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
