)]}'
{"id":"openstack%2Fkolla~999588","triplet_id":"openstack%2Fkolla~stable%2F2025.1~Ib736e5beb201409c8987fbc62225c75f89cbfe62","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"38873":{"account":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"last_update":"2026-08-07 07:27:26.000000000","reason":"\u003cGERRIT_ACCOUNT_38873\u003e replied on the change","reason_account":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"}}},"hashtags":[],"change_id":"Ib736e5beb201409c8987fbc62225c75f89cbfe62","subject":"ironic-pxe: don\u0027t fail if symlinks already exist","status":"MERGED","created":"2026-08-03 10:07:15.000000000","updated":"2026-08-07 12:38:39.000000000","submitted":"2026-08-07 12:37:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999588","meta_rev_id":"41cb2a44a73b5b993a6e0dea60a8772f6f8f46d6","_number":999588,"virtual_id_number":999588,"owner":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-07 12:37:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-08-03 20:31:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-08-03 10:17:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-08-06 07:36:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-03 10:17:06.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-08-03 13:22:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-03 20:31:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"53a2e6f2d24dc92bd471abf9c7718419299dc411","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-08-03 10:07:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb1a46d011f80fe057d9520b9ad8e714a4250968","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-03 10:17:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"252dcf82de0466b62dc8455c62b182b16c3244e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 13:22:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b574df46ffba41bbb279c66427993ed0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/60440586489c4fe8bf5add7716b1d171 : FAILURE in 43m 25s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/441c354b103d4365942739e0f5758516 : TIMED_OUT in 3h 13m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f55fcdec7c8d4e279ae45fa25485bd00 : SUCCESS in 1h 27m 17s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/4e36b2f2a4fa456e8edfcda32ce2a1e8 : FAILURE in 8m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c1d229f33b6946aebf817b8c111be758 : SUCCESS in 50m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f1a667e324094089a7829ff56c6119c8 : SUCCESS in 27m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/124d6a799929475880284da8c0987770 : SUCCESS in 57m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01bc3ccf1afd42ddb3d8423eaa91a0a5 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ef87eb2b705464ba6af6690ec165937 : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a864b23a90a408c8d0c22d3aadfdf4e : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8113b9d963bc4746aa59ee433ba783b3 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/972307f9a70248efbbc2488882641e56 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1524552cc794464da77f31d802d8f07b : SUCCESS in 6m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/21ffce0bd8644ffe8ee6f4f3afeaa005 : SUCCESS in 2m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ef556fc63b86443cbe01f99c705efb43 : FAILURE in 58m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ae0305506f674263bb7ce300fa4a7feb : FAILURE in 43m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9061d4beff6d437a985ac280c412d66e : SUCCESS in 1h 28m 19s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/93340f0e671a43a784bf273863f5781a : FAILURE in 1h 19m 23s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/5114048c5cb44314bbd14db412a9fdc3 : SUCCESS in 44m 43s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/7eecfca9cfa04a9ebeee870344301d8b : SUCCESS in 1h 00m 32s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/ca305f2943a84cb7b3fa374418745a5e : SUCCESS in 50m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a21b0e5e040f4d5cb5479e192904ac30 : SUCCESS in 56m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c584786cc2eb45fb8adbfde87a4c6bd9 : SUCCESS in 15m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/38ed529454454ae78b947ac219318772 : SUCCESS in 1h 01m 33s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/af911c939a764080b6f272e298abecda : FAILURE in 41m 46s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/517731baff014536b2aea35ad3ce40ac : FAILURE in 1h 32m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ce6e3236ae8a8604709500a837d63356e27f6b7","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-08-03 14:43:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1515b7ca5e1f0f6739b27921a5edf237bf9d84a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 15:19:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8630d4f0c8704e949ac55fba98622edc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/608e24f4d8ed4c3594889b0a6b134d65 : TIMED_OUT in 3h 12m 56s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/fa55e16054fa443ca4cea32d14c23d6b : TIMED_OUT in 3h 02m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c6b7d16af04c415d94d0b6ecd095b9f4 : TIMED_OUT in 3h 18m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9646ebc6799f4708afa10a8613c66cca : TIMED_OUT in 3h 02m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3af12c003f964e5ea539bdfed84fb108 : NODE_FAILURE Node(set) request b718e8a4f1a240e98c27d248f535c7e8 failed in 0s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/24344c057d1c4967af3c69549e76d940 : SUCCESS in 1h 53m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9707a42e6071406994a285d5c1c29693 : TIMED_OUT in 3h 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6adbffa40c2a9903d1a7e0ecaa24ce4dd7e2902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 17:52:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/023984b8d1ec4b8e950ee3b1996d9a4a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/758d2738faf74d3eada37f95299db8e5 : FAILURE in 41m 16s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/30fbfd3471fc4f829b4ccf83922775f6 : TIMED_OUT in 3h 07m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8f80a58bec414865abd3ba09bf576089 : SUCCESS in 46m 12s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/e9b21cd0f50044219db591d7f5a609be : FAILURE in 10m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d3d1648c3b744b55afa5c619318ec18c : SUCCESS in 47m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cb0be505811c4c70b0be7d6f60dc568e : SUCCESS in 22m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/39e635e7c1ce46b2a85dec16406bed5f : SUCCESS in 53m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f3461367771457dad4d6d35ddb5b77d : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e123773351234931a44621d15dc493b9 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a03f456d815842db8490da6808af86df : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6dcc51fb59748dd81d45e197ee2acf3 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52bc90719f774f63910a059305401c1b : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a77ea4e22d742fea4aa3727b0812023 : SUCCESS in 9m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a4cced6b4c4e4ca0be24d76282e0851f : SUCCESS in 3m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/53275eeda7174fc69087f53b041d5de5 : FAILURE in 1h 00m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f113c385713e464080ae80ac9006cc87 : FAILURE in 1h 09m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ffa4d27fb11b4e9ebdb3a805695f05d7 : SUCCESS in 1h 42m 47s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/53af5512bbb64e39a94ccb10d4750b57 : FAILURE in 1h 08m 19s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/5b856572a26c49468bec080ab20cfc2f : SUCCESS in 58m 26s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/9528d70a3b854529aef7c7768a9eb038 : SUCCESS in 40m 49s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/65f20074c89244d88481eb55f9763d76 : SUCCESS in 46m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ff2eab46ae614e1186c24c388fdb5998 : SUCCESS in 51m 01s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5682055a0e33401086cb47bb2a99cf21 : SUCCESS in 28m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/15cd111ccf5046bfab2b5ee375f8779e : SUCCESS in 59m 56s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/a1e890f425a548e391af5834b14990b1 : FAILURE in 1h 00m 51s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/0f466cad57a34e96936fb73bd7b12049 : FAILURE in 1h 09m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bffce4f220fdb5f7b44bbcd7d94980b2cdb7d054","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-03 20:31:07.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a1f39303da8a5e4553757a0ef24c93a53ebe52a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 22:54:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5174c258b894485a97e5c906f643d386\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8b2d4ed772374ebcb922e2696744a8cc : FAILURE in 2h 05m 58s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/7b6927e3423f47dd9ba8d182f5c27196 : SUCCESS in 2h 21m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7dc4574780ce4fd48677695a2cdc5940 : SUCCESS in 2h 01m 13s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4cfa2b29f98f4ecda9471d0cdf6f4adc : SUCCESS in 1h 25m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/29b29cdf07184065b69240ff5ad03327 : NODE_FAILURE Node(set) request 348fb93c9fe54149a6d27f7e76044e91 failed in 0s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/4e25efeefb3b4260adaea5599cf11b62 : SUCCESS in 2h 12m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/90c404ef92804d9e96f98a761c053f17 : SUCCESS in 1h 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"127803341f8ae5481795a1524a03423b7a796d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 23:50:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd4cd53a567641c395817b2a0bdd309a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d66251a6369b4015a6a2df7fcfe97c9d : FAILURE in 42m 29s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/84ac72629a4e46e8a4c93007a9670341 : TIMED_OUT in 3h 05m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/201a06b50db7450a9002d1f13892ac47 : SUCCESS in 1h 14m 36s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/27c0893fcf9b4a9a91897a952f1bd2ba : FAILURE in 8m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7b0f7567a6ab4bf4a87e87aee3f6e09b : SUCCESS in 56m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0165ccbe218849d398faf9bc12062934 : SUCCESS in 19m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0bc511a40cd545c494d79921a751fda2 : SUCCESS in 51m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13450b6e33d949b3abf86a0435872e2c : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c09e19a262442168fd27d4b3cc57bcd : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9bbd45b1c2c8451492c9463646099a4e : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9c92c3fa738404d9a0f8f804373c7db : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcbe8a50bc3b4566a0d890d9a50845f0 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2defe634c2c847039679b17bb5bcf743 : SUCCESS in 3m 42s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f363d3d2da8e4699a9d318de2193300e : SUCCESS in 2m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/41c8c1276dac4d74afd532f0408fe7f7 : FAILURE in 55m 26s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d03b084607d24710a188642da427c587 : FAILURE in 1h 20m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/be699b11b80e40799645795854c718a2 : SUCCESS in 1h 28m 19s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a0e2d35c54374c9b85115c8dc3fe4c24 : FAILURE in 3m 53s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/2f95584c42db4561ba4c266b04e1612d : SUCCESS in 56m 31s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/696ef37d0c224c47ac816d66213a5246 : SUCCESS in 1h 02m 49s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/7efa0210e7374341a7888fcfa5db021a : SUCCESS in 1h 03m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7e96dc6a33b74f50bdaeb49398ebae83 : SUCCESS in 33m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/912e30471a4546199ad1e2bdf5a81cdf : SUCCESS in 36m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dc40b72be0ea4d42a017224701b96e80 : SUCCESS in 58m 37s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/87c1721bd6d544e2a02d2257cddc65d7 : FAILURE in 1h 05m 43s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/0ac115a76d494d428d8f1517c9416f81 : FAILURE in 1h 35m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd551a3b60918472410523406f0541e8656b0fe3","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-08-05 15:51:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c46749a3fe4876226c004cf5a7a8868e666dcf2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 17:34:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4140dc30d49246cd8fa2b0d4af3ff2bd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9f7f1fac83584e5791e0404bbbc7c90b : FAILURE in 53m 46s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/996a4400de8f4c6898faa1c420861e7e : SUCCESS in 1h 27m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fcad4099ba45466f8bc7f5acb5582cb4 : SUCCESS in 50m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/808e7b7766c64b5d962dd299d94c5595 : SUCCESS in 1h 11m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/266325f40197426d89af99d7e899fa91 : NODE_FAILURE Node(set) request e99503f7ff7a4bdf978cea0420e4bc29 failed in 0s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/aa4ebb9d0c5f4f89b3719764832ead55 : SUCCESS in 1h 17m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b58b7cf5e49f485ca3cef5f31f7f08e4 : SUCCESS in 53m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8939ca195d7f7d571255287cba1ce928e6d8e252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 19:18:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86a94e9a827b43a68a7c692ac6f056ca\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5640ca3999ad453e8e0fd27b186cf6bd : FAILURE in 1h 04m 17s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/6c26f9eb957843f2aefbe169b7d38ae9 : TIMED_OUT in 3h 04m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/63936d26b32c45c5837566eb7a98a185 : SUCCESS in 1h 04m 24s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/3f826aac8c7840ec8cc902ceb4ef561c : FAILURE in 8m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/892374bdd3554e01b3418900b4767dd1 : SUCCESS in 55m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6e2c86116b3f4ab0a3d9f71f986ad783 : SUCCESS in 28m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b8e0072d16394f298d3d8f221904f767 : SUCCESS in 58m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd1a37ad04794bf7b748384d29122065 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9cd1fdc35854998aef62df03bdd6882 : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/899a8266f548452191f8738e362d06da : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b9ce8dcc5b6437f9d9b887832a788e2 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca8e77a5fb54a66957f1d2b65991ccd : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9c13b40e27146bda9030c0987caffda : SUCCESS in 10m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/76406db5b344470d985975c61c302b21 : SUCCESS in 2m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/33e9e64172794b7eaff2a0111eeb1c9c : SUCCESS in 1h 00m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cb3db9cb51834dc9b2bdc784b1afca59 : SUCCESS in 1h 31m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/22718456c26f4875b311da52cb8573e1 : SUCCESS in 1h 27m 09s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/da0c67bd576448f7a53292f07f77a8ad : FAILURE in 1h 09m 02s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/8e06f62ee52b4e449de0a89379c58d15 : SUCCESS in 1h 00m 23s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/ccf474a7e51b4d3dbf28017c0bb6075a : SUCCESS in 1h 08m 09s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/0f77f43918c9447e849fc2eb0a4ef222 : SUCCESS in 54m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/83ee6ed2ec9341c198b4802120c9053a : SUCCESS in 59m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/043ec6da7fcc423ca26f22a550e766f6 : SUCCESS in 35m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/59517f0cbe624572973ef10738fae7de : SUCCESS in 1h 09m 09s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/94b0c50da3d545459896f461dc556f24 : FAILURE in 1h 13m 41s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/1402b02efb864916b53517164aaf4ba8 : FAILURE in 1h 37m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e158e2d3c30fd756aa00ab99d63233e40f344d49","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-06 07:36:05.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08c411c5c72aed3d6e332cbf546f78d560f22fdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 07:36:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ed5274f8350af81fd9c3bc9c91824328eabb99f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 09:03:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ec38aa131c24207846fbdafbdb6cb7c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d4bd5df532004782b18194844efd2947 : SUCCESS in 58m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/078a0a4845f9458e84fbc725aad8b9c9 : FAILURE in 32m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ba5bd1c47e5040b0ac55e90b58262ed6 : SUCCESS in 53m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8fdcbce1f2d4215a9fc3e49dad8c5e2 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b763a5c9aa746f292eef31d34f84a4d : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/510414bd437b4dcb82ed7b0396786368 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e06749559d4409b0574cf3ca3157e3 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6de8e5fc6040417cadc07cb00e5226fd : SUCCESS in 8m 40s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/96e742b56a4f421185ec13b5d9d90c1c : SUCCESS in 2m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7253bbeb8f76404681a48b276c979513 : FAILURE in 1h 02m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/161c6c9789e14b5b9139e4213807796e : FAILURE in 1h 13m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/27950410bad74c2baa5c03b815f60538 : SUCCESS in 1h 26m 30s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/40b89be371174ad4845e130eaf8d11ab : SUCCESS in 50m 47s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/d422ee2453634f7fbb98cdbbe6b0bc76 : FAILURE in 56m 04s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/c810980f6d894d6f8d4762e88c15a763 : SUCCESS in 52m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6701a17375e24af2a62fcf5c619e89b4 : FAILURE in 50m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7300bd35b548414b9db851823544c123 : SUCCESS in 20m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7dec0f3d878846dba54b6e505fd04b67 : FAILURE in 22m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e6e745feff2f9f795c523ff2608f4b54177286b5","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-08-06 09:03:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1e9be70f7f8e75ec901f3310562e293257ef27d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 11:12:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4d142001945449d826433d1afba821a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/05ac820bd0b843b3a1835166792b0751 : FAILURE in 59m 38s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/66a12c9cad0a4f18ab780e7c5d5b2103 : SUCCESS in 1h 07m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2eec3d2cf1f14aeea0e90806ca3e5cb5 : FAILURE in 1h 44m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/172bb8e6721e4b409cab011e4d159044 : FAILURE in 1h 02m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cb8bb502f9364055a9ba97bfd52bd12e : NODE_FAILURE Node(set) request ac3e53476c9d4fc1ad76dee7f3607b27 failed in 0s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/6a59771c4c99438cb92d3b426bf99250 : SUCCESS in 2h 02m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/73adffd278fa4b71bbac0c40d5366ca0 : SUCCESS in 1h 06m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0dcbfa47be6aed16931e1a16a8ff931b422ce10c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 12:50:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfb32f44432c40fab0fd8f1a579ecd2e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f3e72c2abc084f86a6de5b9d985d3a80 : FAILURE in 1h 03m 09s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/89b0e69cfbbb4e958bc138a6a6c3299f : TIMED_OUT in 3h 05m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ff55234400504346935b7de83c06866f : SUCCESS in 1h 37m 05s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/03d47edddda24053b0ea37a33aa9eabd : FAILURE in 14m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/482cfebf92d94dfca950e395f0d427e7 : FAILURE in 52m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dae9c5ea5b224616a1bddc955cceb47c : FAILURE in 38m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f3a000b78c494eaa8fffacdaf91c653c : SUCCESS in 1h 00m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/745e56e457244cd89c237a88f638283c : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3425a3dd8114f3c927615b5f05b6635 : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d59d8a89695148d0bf8844edc4fe2c5d : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f27c7870fd014ba2a692034ffee74c43 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f7c7f6bd6e47aeab82056f72d2c751 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87995fd299ec42ca9e11b1cd1a8e0c83 : SUCCESS in 9m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7ebc840394d8488daad920a761309278 : SUCCESS in 3m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1e3643ea3112426788783ab35b10cd6a : FAILURE in 45m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b1f56399aa8f4aaf9b9060ad36c950dd : FAILURE in 1h 14m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/64a5988312204a5a95ef663af5c259b6 : FAILURE in 24m 12s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/0e7adffbc1f6450caeb324530de462df : FAILURE in 1h 27m 20s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/9d0910288cef47babe02019674dc2994 : SUCCESS in 45m 39s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/192270450d6143c3af7e943fa0e62885 : SUCCESS in 41m 31s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/fa7b366123e74baaa4b18d8f7dfb4087 : SUCCESS in 48m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/11b6a488d26b4319a79518e1b62d1af5 : SUCCESS in 35m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e1985a602f9140cea337fa7f05ebb7b0 : SUCCESS in 29m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/48e49a9c060247bf8d1486fdb758942f : FAILURE in 7m 12s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/54e06765085045caafd1af1276f52a4e : FAILURE in 1h 17m 12s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/59b49efb1c5644aaa89a93b1fc3703e9 : FAILURE in 1h 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70d8d7486d510021bcc4526592b9565db24c873e","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-08-06 14:15:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9c5f932c59117b679ffa0ec07b247d2039908d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 16:38:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef7500f982f947f4a4b248162e8de721\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/596c2bb358434f4fa4d1ef2640308a76 : FAILURE in 1h 05m 05s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/9bda8dc3e0db4a34b33be093a224ee4d : SUCCESS in 1h 13m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/00f99d9a91254e48bf1fed53334510aa : SUCCESS in 1h 29m 15s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f463ef8e3025486e832c4927ee1080be : SUCCESS in 2h 18m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/88109da74fb045d8b90256950bc165a3 : NODE_FAILURE Node(set) request 10f0a64f08664bebac49bbc406eb0350 failed in 0s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/66794239aa264b64a8c49c46bd154a88 : NODE_FAILURE Node(set) request 409c9611ce6046a3a49eccd43cf346c8 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ec20469c0f8748caa33253f866e19c70 : NODE_FAILURE Node(set) request e2d5bcbbfd2c44f495b9c4c2d27e795e failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa0e0f6c41f7621121c74577a791dcebfaaf8eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 17:27:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72b935c6b871474da61c75d2bfdaa402\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/593cc46092ff4513ba079ca4fa09ef22 : FAILURE in 55m 39s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/bc7b577b0d78491c86a4f4278883a98f : TIMED_OUT in 3h 08m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/21f9c07fefe24375abb6222fe78169b3 : SUCCESS in 1h 03m 49s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/00a01ec10ed04367b4775c7036bc5e9b : FAILURE in 13m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/136a6299ff414416ab54ba78660ddc82 : SUCCESS in 44m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4106bc4f03b5477ba5315ab32f67167d : FAILURE in 31m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f2899dd95ca84787ac8e9a4b09003a40 : SUCCESS in 53m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cec0094bcdc43d49877289781e449ab : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6393c8ed238548c7963acd659756b98b : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f972e486f9c64105a31e45597263f2b0 : SUCCESS in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b3db4e4f4254effb83e17a828aa34f9 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1692f2d3cc40569d33aeaa8f4a0622 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b066331c27be40158fed953b648e8ac3 : SUCCESS in 8m 53s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7f59f8cccd9749bfa0819ce71e7302af : SUCCESS in 5m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1f2886c008cb44568570c3743e928c4c : SUCCESS in 45m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fb561d210b814219be1c3b22292e8602 : SUCCESS in 1h 09m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/52ef2e9f1c8d48dda4396831cb3e4c77 : SUCCESS in 1h 33m 22s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/73b486590e16411ba66fcf02ac829db8 : FAILURE in 55m 34s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/e4dc5121795543bb92d215acc77b5c85 : SUCCESS in 51m 20s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/2f2b6b08dcea42308155251a80b75a60 : SUCCESS in 1h 13m 06s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/b5d48e67492e452891ce73b202d166b4 : SUCCESS in 1h 12m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/64131201cc5945c7971f275734c1588e : SUCCESS in 1h 01m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8991cf722ead4f01855d2b0ca62a3d79 : FAILURE in 37m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/18d54eb615cc4d22b30f1c94929b52d4 : SUCCESS in 1h 01m 38s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ce3747699b06448a8c8b2832c923cfe3 : FAILURE in 30m 37s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/254750a46a584cd49ec8801b6662c93e : FAILURE in 1h 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ef143a45bb60a1a16c4cd54f49dd17239012fd0","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-08-06 18:07:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73364707b1c30f3fea6d309b2b8b648c95f1897e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 21:14:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b390d5c299d44c9a50086f6e33d1fde\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/eccfa85d7def41e5899c6d0681263455 : FAILURE in 1h 19m 13s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/fcc20c7a5f074a328ffb536815bf4375 : TIMED_OUT in 3h 01m 40s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d1e6089148294927a2e787211e5564d4 : SUCCESS in 1h 28m 54s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/6c7d0db90af9472e880795e0b5bf4b4f : FAILURE in 8m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/78950ce89dbb49c5a375edf608b42886 : SUCCESS in 43m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d207e03856794549931f8516b85dd25c : SUCCESS in 29m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/19e7222700104ba38c5d9705669bf6ed : SUCCESS in 57m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a398f3e217334857bb192f9f1f4095b4 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75a2071cc09a4859bd9861c08ede854e : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28ca125030f24dd180acfc6f78556ba6 : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f47e64b907824fcaa8884a227445e2b7 : FAILURE in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a32cb15b4e1f4111a358c07b581da018 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03b9fabb4f9f40e1b0a743291ddd804b : SUCCESS in 10m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d5c15859bd754939967d88a75a2eee0c : FAILURE in 2m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e1a7c570e4f3490bad8f2de4d42d6c92 : SUCCESS in 1h 01m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/29c698e14efb4bad9ac6b9372f277bc3 : SUCCESS in 1h 18m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5b7b4581099a41b6aa9bdc16c88a91ad : SUCCESS in 1h 34m 38s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3369db15a93f41eabcbdf24b363993ef : FAILURE in 1h 44m 23s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/a11b345785ec4f74ad66816b9212b922 : SUCCESS in 58m 25s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/edbbcd45a7494c49afeb9817d6e489a4 : SUCCESS in 1h 14m 53s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/66db3da51201429db7e80ec4ac450993 : SUCCESS in 1h 04m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/de3e221788454484a6448f2884c29702 : SUCCESS in 49m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cf1d179761b74c54a088ef442c4e46a3 : SUCCESS in 39m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/145df1fb867e4cd192df34c055228ba2 : SUCCESS in 1h 01m 22s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/075fc2f6fdd14e11beccc4957d5bf586 : FAILURE in 1h 15m 39s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/f3d1e671278443cf943cf9f2c85cfda2 : FAILURE in 1h 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14339893a8fc1695ae00b13af96cefcbe6e57388","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 23:46:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9792173feab443a7ac8953a6541d7a31\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5c43a1bf2af24a9db332317e06c6ba97 : FAILURE in 2h 21m 36s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/cddd299793744f028456f66fe6f7c42b : SUCCESS in 2h 43m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bdb427a92ba2403fb8716a7da946744d : SUCCESS in 1h 10m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e5c6d4681ae743bca25696ef35615340 : SUCCESS in 2h 52m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/04c5a1b8368344e298c82e8f3ef17425 : NODE_FAILURE Node(set) request a262d9d732a64bdab6ea4e70d69bb8a8 failed in 0s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/643938f212f44989bd95115eb68020f8 : SUCCESS in 1h 45m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/96159cefc6ec43d7810e3535c01ab764 : SUCCESS in 2h 55m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21bd25660f01cd80f5d48d31624bfa8d75a8e383","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-08-07 07:27:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff274b8b55b2885e895b437b639afe9c536cdeae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 10:24:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59765bd9a43348fdbba126881e8381cf\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c59bc9bbad804b3b8a0473fb87859ec3 : FAILURE in 1h 34m 53s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/def655d9cf1843cf8ca8d0051055468e : SUCCESS in 2h 54m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3238bcdd6b5b49f983ad50db3be69189 : SUCCESS in 1h 23m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4897732343a44d74af8f907c2c9acea2 : FAILURE in 2h 05m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/eb9daa513cd64a88a68f19ee05f38c29 : NODE_FAILURE Node(set) request 08d383cfc6984ca5869e36c2d16ea366 failed in 0s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/4e4b4372742a46b69cf68c0a8b3dd540 : SUCCESS in 2h 48m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d7270d2c212b4ca4bfba3b3c28bf51f2 : SUCCESS in 1h 37m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4246ffa3544a3acf6de48537438ff28dfc8a6dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 10:48:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/598cd122c6cc4c4abb72a6559a4eb692\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/62db272e505047a5a8a6fb3e2d4a744e : FAILURE in 43m 16s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/18e6cefbb21847abad6accf2f05c978c : TIMED_OUT in 3h 08m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/590a40f129a54bccad819dc690318491 : SUCCESS in 1h 08m 43s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/6883ab5a180c4bcc9f5147e7ebd6bd85 : FAILURE in 9m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/260870fec51e4445a720660ca57529a3 : SUCCESS in 51m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7cec8a72ccce465aa02ac0590fbe0275 : SUCCESS in 28m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/23e4a111d4f945438a814647064191e9 : SUCCESS in 55m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66a5edabbc6e42ef8d1b8f60dcdf15a1 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a8afd720cad442aa7a13581c7bbfaf7 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c799393d6dde4dd1be5fe39c4f16f96a : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/712b9dfeadda462aa82cb83d4a2d5d49 : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3160d6ebe3c94e8c9c0eb178008c6e3a : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/410858eb269e43108c58d0fd41f38069 : SUCCESS in 9m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9980dd9b16d74101ac1cbb8cbba4e849 : SUCCESS in 1m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ac5250440bf84a0faf34acf53c5401a0 : SUCCESS in 47m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/64abe8d9a9764aa0850479de0abb6a2a : SUCCESS in 1h 02m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c4f46a9237574529810718077c91d1fc : SUCCESS in 49m 31s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/4738a6721d0b4b8b94d17eae27ee97a7 : FAILURE in 49m 29s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/bcc2f1baaa264fe8a44c7f86da70abad : SUCCESS in 1h 00m 10s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/387345d452db41f1975a3071bbe0d811 : SUCCESS in 47m 09s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/dd6b164402fd4a0ca2c02f77d7c3d6fa : SUCCESS in 1h 16m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/192d593b60174c75a5654e1d76781be3 : SUCCESS in 36m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2ee099df5abf424d8482eaefdd0e827f : SUCCESS in 17m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8ba76408a69140579a965115771520de : SUCCESS in 1h 01m 12s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/b5a52e777aaa424cb9df0696e1b33abe : FAILURE in 1h 18m 41s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/43beb64daf5f44379202c62a2dce8242 : FAILURE in 1h 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90f1432c20c68626cfd87295e1c970741cbb847a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 10:48:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e0b3718cf7cf32211ae684054b0b238e75a2854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 12:37:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0268ee3698a94ddca60db457c9c9ca4a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/975f2e6ac5524942bdb405ab12a604a4 : SUCCESS in 51m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/97f2bd834c3c448e9cd5268e90837823 : SUCCESS in 31m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ddf99abc317a4629a75e3220bed26a53 : SUCCESS in 51m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d64aa1d333e740f7818b04848d4ba16f : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b89368ff75e24da080925ce7a67c21c9 : SUCCESS in 6m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3bc72f0ed711493b90cef8368f9f4c7b : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f5f5aaac4f438eb0307e47ce71267a : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21b0ecfe2bac46c39e4e2cd26d9f94ed : SUCCESS in 10m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/be09248e3ae94c948154880c4dd69934 : SUCCESS in 3m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6b1f6813341b495eb2c9eea56e2a3d7a : SUCCESS in 54m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/181151992b36474087ca32d8ab5fb783 : SUCCESS in 1h 00m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a2b3b74e61c848baac5826e1d5472e79 : SUCCESS in 1h 34m 56s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/256ba7d6452b4ffbbc94dd0b91af542c : SUCCESS in 1h 02m 51s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/bc3b41e840894a71bc3171d2a3d18aa6 : SUCCESS in 1h 14m 56s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/a85e0361a0154a62b734f29f6dad7446 : SUCCESS in 1h 09m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b626ddb8d5d44720a9ae3eb648a01a50 : SUCCESS in 53m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ddc25fedbb3f4a8dbd80d5753ab9a5bf : SUCCESS in 23m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9e9f468dd8fc48a881eb806844ac9497 : SUCCESS in 58m 46s","accounts_in_message":[],"_revision_number":1},{"id":"2fa6a9861fe4817e52f03c2a1309e80084aaa971","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 12:37:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"41cb2a44a73b5b993a6e0dea60a8772f6f8f46d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 12:38:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c2b182f7b1249e392b2e40af9f80abf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12495f8ad19d4306b73509202c2af5e3 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dc36ce5bb8d42d488a22eb5a899f295 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"547d31c6067ba6523e17cd8257a8e15bef532e11","revisions":{"547d31c6067ba6523e17cd8257a8e15bef532e11":{"kind":"REWORK","_number":1,"created":"2026-08-03 10:07:15.000000000","uploader":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"ref":"refs/changes/88/999588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/999588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/999588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/999588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/999588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/999588/1"}}},"commit":{"parents":[{"commit":"d1c4dd49b92e68509a413c33667bbe87cc3d3a9e","subject":"Merge \"Add genisoimage package to nova-compute-ironic\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1c4dd49b92e68509a413c33667bbe87cc3d3a9e"}]}],"author":{"name":"Owen Jones","email":"owen@stackhpc.com","date":"2026-07-10 13:53:24.000000000","tz":60},"committer":{"name":"Owen Jones","email":"owen@stackhpc.com","date":"2026-08-03 10:02:21.000000000","tz":60},"subject":"ironic-pxe: don\u0027t fail if symlinks already exist","message":"ironic-pxe: don\u0027t fail if symlinks already exist\n\n`ironic-pxe` start-up fails if symlinks to efi files have previously\nbeen created. Removing test for existence and adding `-f` flag to\nsymlink creation ensures they always exist without failing on start.\n\nCloses-Bug: #2160317\nChange-Id: Ib736e5beb201409c8987fbc62225c75f89cbfe62\nSigned-off-by: Owen Jones \u003cowen@stackhpc.com\u003e\n(cherry picked from commit 2da1223d8301da05f9b070ca656d62c1bde93c66)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/547d31c6067ba6523e17cd8257a8e15bef532e11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/547d31c6067ba6523e17cd8257a8e15bef532e11"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
