)]}'
{"id":"openstack%2Fkolla~999697","triplet_id":"openstack%2Fkolla~master~I7761f689c761ffae051d90f9d1441b3e355fb466","project":"openstack/kolla","branch":"master","hashtags":["kolla_openbao"],"change_id":"I7761f689c761ffae051d90f9d1441b3e355fb466","subject":"Add openbao container image","status":"NEW","created":"2026-08-04 09:05:20.000000000","updated":"2026-08-04 10:41:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"ae4fa7fbbfae2a7eed4c7f4428594c376c3893fe","_number":999697,"virtual_id_number":999697,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-04 10:41:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-08-04 10:28:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-04 10:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb2c2bc923ad115ab5e2531484d0f22a5df744bf","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-04 09:05:20.000000000","message":"Hashtag added: kolla_openbao","accounts_in_message":[],"_revision_number":1},{"id":"b6d68ab3b4201c58e94ae8bf6fdd4a21004edb66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 10:28:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ca8a238adb4486da208ec65b0ab15da\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7a099860a4be4355b0a64bce1d2c9b3c : FAILURE in 41m 10s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d110e91c9f91425595ae38e046ae344f : FAILURE in 35m 32s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/de9e54bbc0994373af336f7c9bd78693 : FAILURE in 42m 37s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7bc8c04bd923494e8a8a9b1df2cb1f19 : FAILURE in 37m 16s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/a62f6f26a27246bb8270a9f9ecc670de : FAILURE in 1h 21m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9a4d154955bb4625908ceb90a679937a : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fa764155ca6a4c0aa6f8282e3fbbe0c0 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d1c1a8828caf4ab7ad0d5b2788c7297c : SUCCESS in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae4fa7fbbfae2a7eed4c7f4428594c376c3893fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 10:41:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c115a60cd5c47a983f62eeac7d708f0\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/fd4ba859ab8c4ced850945131b618fbe : FAILURE in 41m 27s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/71d1e42633af4bee9f1645039261e77b : FAILURE in 24m 46s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/49e4a7e4a3bb4b56ba07855b9579d145 : FAILURE in 26m 27s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/6e6a52ab3700413f96755e53bdacddd1 : FAILURE in 19m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0dd32969531e49fa82c27192a3c56bcf : FAILURE in 51m 00s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94f030aeeb3549c6bc46e68fd7d1cba1 : FAILURE in 23m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/89b84afe364647afba3577f1b0a0f357 : FAILURE in 27m 49s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a9daaa2d313e4c11acdcaf44b1af4374 : FAILURE in 46m 32s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/59c8b63d964d4c7c8394abbe8ea18b23 : FAILURE in 1h 34m 09s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/450fbebdfdf443319e53adbdd89216d7 : FAILURE in 50m 40s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/aa91243a3c5e42e1a95dac72e47d4927 : FAILURE in 1h 34m 33s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/585006651db243538aebfc7970ea3e04 : FAILURE in 49m 24s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/9f657a1e3c724ee1a1d0e02f1631850f : FAILURE in 1h 34m 11s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/859865ed4c4c49118773a3425c4994c8 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dfad59dcc324a278f9cb79e67ce13b3 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31ef5562a55e41f09adc1bcfdfacc2a7 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b90f6cb4d124b2c8e4accf4a75ede26 : SUCCESS in 3m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/be7d36250d0149ecb36c41ad3e58c516 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b303dbe8de944729d94dbc409b35daa : SUCCESS in 2m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01cf4f59f22942c4a48b5fdc7f055186 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b071c31a0bd4e3496bad7d0740d2e21 : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79e90ef1a582a03cf790826a002e4d52355164c6","revisions":{"79e90ef1a582a03cf790826a002e4d52355164c6":{"kind":"REWORK","_number":1,"created":"2026-08-04 09:05:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/97/999697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/999697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/999697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/999697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/999697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/999697/1"}}},"commit":{"parents":[{"commit":"8e912a9db7f013a1abfc1e70a829e435d50622db","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e912a9db7f013a1abfc1e70a829e435d50622db"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-04 09:04:52.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-04 09:05:18.000000000","tz":120},"subject":"Add openbao container image","message":"Add openbao container image\n\nChange-Id: I7761f689c761ffae051d90f9d1441b3e355fb466\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79e90ef1a582a03cf790826a002e4d52355164c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79e90ef1a582a03cf790826a002e4d52355164c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
