)]}'
{"id":"openstack%2Fkuryr-kubernetes~689421","triplet_id":"openstack%2Fkuryr-kubernetes~master~I4122d73ca6e653b11c856fc7d3576d5e8af496fd","project":"openstack/kuryr-kubernetes","branch":"master","topic":"connect-failure","hashtags":[],"change_id":"I4122d73ca6e653b11c856fc7d3576d5e8af496fd","subject":"Catch right exception for ConnectFailure at activating vif","status":"MERGED","created":"2019-10-18 10:43:48.000000000","updated":"2019-10-25 06:38:03.000000000","submitted":"2019-10-21 14:48:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689421-1571669313695-fe29816b","meta_rev_id":"a96b2101b76b01099c42ecbf50a84530f8b68091","_number":689421,"virtual_id_number":689421,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-21 14:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-21 11:14:44.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2019-10-18 10:54:22.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2019-10-21 08:22:43.000000000","_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2019-10-21 08:22:43.000000000","_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 10:54:22.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-10-21 08:22:43.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2019-10-21 14:48:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e19c36715a955587f5570c2a7f0b1a9d0ee1ea8","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-18 10:43:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe71d8062f20b16f5b55c4df9f0644d1923c143e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-10-18 10:48:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0869d1136b2aca332f5a6675224c427de8485fe5","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-18 10:50:57.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"53e278856897cacbdf31066491b92e011cdea44b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-18 10:52:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"058fca60738506ad3d4c3d0d37c91f5f41329d35","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-10-18 10:54:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d5145c1fe0e527ee708b48f95a9274b45b1819e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 12:31:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52de0df1d77549cd8fb7ca2cc3d2b1c0 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/37eec24d66d046338896a56cf8bdad53 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b00cf9bee264108947196547f4aa3e0 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/735aa329fd9f4ac9a2695065d6d4f288 : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e06ae051ef043c4a150a309ed0e7856 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d6bdfe44324796ba94b93ae2f8cd1b : SUCCESS in 4m 50s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f504e301e3ed4a278f2f4b05d2fdc09e : SUCCESS in 1h 15m 32s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/43ef2e11bc544b10ae257c997653fec1 : FAILURE in 1h 33m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/ea18dfff8723424cb774aec9c1ec09c5 : SUCCESS in 1h 20m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/f0621cafcf674599a926d69acb715e99 : SUCCESS in 1h 16m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/dcb161b7f182439c8b20fb90b20f46bf : FAILURE in 20m 27s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/4084e8cda3584cd7abef4c12a34e4bc8 : SUCCESS in 1h 10m 06s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/f5f012ca3bcb49539ebe9a85089b873d : FAILURE in 1h 06m 25s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/c969b373184c4cd0a47b93a9c77cb614 : SUCCESS in 1h 31m 50s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/eb507a231e3349979a9f04dfc45c80c4 : SUCCESS in 1h 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"55d93c51d759bcf8c9f36e37e8175f81c3b126d8","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-18 12:50:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4c411231e09fda4f30cfd24a7a4b65fa7056b3a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 14:47:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3be180a35ad04fc4836bd251a28b817f : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d181cea18b3470290357001b95fb4e6 : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed9978ce84ef48cca35168679b964799 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd0d62931f8946b69f333a6b2fe2b6b2 : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83fb277ebe114d54872d0184cc4c8208 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6b4a8e51688439087f0a9f2c2599819 : SUCCESS in 3m 20s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7981d43e245242ba98ace10d9ef9d514 : SUCCESS in 1h 07m 54s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/aa1e3c367a9543f780fcb7435ee4bdb8 : FAILURE in 1h 22m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/c7d6db7ccb674c99a2f284b060c348f5 : FAILURE in 1h 03m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/d5826f329c5f4049803496228053bd70 : SUCCESS in 1h 37m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6a3e20ea1a034daab1765b04bfb019c7 : FAILURE in 19m 47s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/7473d4c7e4d64de6b7114e8d548f5370 : SUCCESS in 1h 11m 11s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/b1ee6f03a4db48fc9b90db6ce733a2a3 : SUCCESS in 1h 27m 21s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/7caa74a91cc64cfda64d2612ba01cf22 : SUCCESS in 1h 52m 11s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9b855fa454ab4e7fb0a05a94ced9fd7c : SUCCESS in 1h 36m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b18ab7b9907e17849825fdac141f714e4851f144","author":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"date":"2019-10-21 08:22:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aac851df0b6cffcb6e49745aa68e1f6465ca2636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 08:22:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2712ea57400c4791c89f3df7a65e5f9d1fc81a57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:13:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81d00bb4103446639c289ceedfedfa83 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8832b093d70c48908b38bd409e3c5ec0 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/198c844e68c44cbe8f2d6558db4d809c : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a14b5ee7bdb4520aea81975d40ac35d : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20417d83267d41f290c59c7de5d6b69d : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68fdd46d34fa4f22a215268926caeeb3 : SUCCESS in 4m 08s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9f90c40de8d84572861a0544caf0be53 : SUCCESS in 1h 09m 41s\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/ffc7fe4bed2445119f42033a00dc56c5 : SUCCESS in 1h 15m 09s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/91654d91d068428ca579b4f0cd7eba92 : FAILURE in 1h 45m 13s","accounts_in_message":[],"_revision_number":2},{"id":"8cd5259b24a4ea806387fca76980d100bef2367e","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-21 11:14:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"18a8cacbf46add2c300a31db98ee922c271547f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 13:15:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/037eaa5a05de4ebb83fd078655cc8bb0 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8fa63bd2c61b4fbab31461cb0827fbb8 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23a46625c759458aab59a3c29a367509 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd58bc4eb63a48e2b14b0f3bfc66a15b : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8ee3179c62c4aa19dc4e1d15f0c3b05 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce7a3c77d1847a1b65582375d7078e5 : SUCCESS in 3m 42s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5764cc9bfd3f41459b8a37becf0b3f3d : SUCCESS in 1h 37m 44s\n- kuryr-kubernetes-tempest-openshift https://zuul.opendev.org/t/openstack/build/b63cd0fb69b943ad8896abe82dd736b6 : FAILURE in 1h 56m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized https://zuul.opendev.org/t/openstack/build/e66ae70d670c4d26b751eca8d2149102 : SUCCESS in 1h 36m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints https://zuul.opendev.org/t/openstack/build/77bbdb7ba1db4ca8a1ea20bda06bb7ca : SUCCESS in 1h 21m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn https://zuul.opendev.org/t/openstack/build/6a90515afc544443b6caec46a88f0f98 : FAILURE in 20m 40s (non-voting)\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/a19c67d2a7c54925a91c557387f19c53 : SUCCESS in 1h 15m 31s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/b85623c388194be18b0791feb38a6abe : SUCCESS in 1h 46m 37s\n- kuryr-kubernetes-tempest-multinode-containerized https://zuul.opendev.org/t/openstack/build/bc835fb207b54774826f0caddddbbb82 : SUCCESS in 1h 33m 31s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/450e8082ddc14f02bb5ef2ae4e28e3e5 : SUCCESS in 1h 34m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d1a766f2577e9c6ce3dcf33b71cbe4cb2646edd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 13:15:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"730ab2c2943484c65505f94506302b59460a1c4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:48:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25e04d1eece64ac18cbbbf5fc9d09a9d : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab9b22bca40041159ff1a3a6abe85831 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/320edf2112044ca8affb25f826fa0004 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/367c036b5882481e84c6117046858fd2 : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e0072ed175043f98aa029a1b8b7717b : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/817657ad16e4494b9d045ad6a47e73e9 : SUCCESS in 5m 21s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8ed5edf7ee074a4a86fec1f960a7bc77 : SUCCESS in 1h 16m 12s\n- kuryr-kubernetes-tempest-py36 https://zuul.opendev.org/t/openstack/build/1532572a8a8d406087fe699ceafd45de : SUCCESS in 1h 23m 39s\n- kuryr-kubernetes-tempest-containerized-py36 https://zuul.opendev.org/t/openstack/build/b416aceba0b84cdb8d48dec613168dd9 : SUCCESS in 1h 27m 01s","accounts_in_message":[],"_revision_number":2},{"id":"bd47de9d5416b656b0531dad1de71c9389bf5d58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:48:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9eaabce6cc2b6deb72a726d7299fdc15dc0d2247","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:50:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e8abacd5b84529aef4afa2ad486614 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2},{"id":"861e59af413ad574f1325fa8e8f29eec53958d2a","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-10-25 06:38:03.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 8a6009d0e4bed7ec10828eb978716cd7da0174db","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d4427604428e3fd69c5728a6fed56359b1889d8f","revisions":{"c36692881efaf5bf70d07437b877758e70371ecf":{"kind":"REWORK","_number":1,"created":"2019-10-18 10:43:48.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/21/689421/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-kubernetes","ref":"refs/changes/21/689421/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-kubernetes refs/changes/21/689421/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-kubernetes refs/changes/21/689421/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-kubernetes refs/changes/21/689421/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-kubernetes refs/changes/21/689421/1"}}},"commit":{"parents":[{"commit":"6446f70da0dbc9f69421562c837c487e36fcfc0c","subject":"Merge \"Just check k8s 1.16\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-kubernetes/commit/6446f70da0dbc9f69421562c837c487e36fcfc0c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-10-18 10:42:59.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-10-18 10:42:59.000000000","tz":120},"subject":"Catch right exception for ConnectFailure at activating vif","message":"Catch right exception for ConnectFailure at activating vif\n\nChange-Id: I4122d73ca6e653b11c856fc7d3576d5e8af496fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-kubernetes/commit/c36692881efaf5bf70d07437b877758e70371ecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-kubernetes/commit/c36692881efaf5bf70d07437b877758e70371ecf"}]},"branch":"refs/heads/master"},"d4427604428e3fd69c5728a6fed56359b1889d8f":{"kind":"REWORK","_number":2,"created":"2019-10-18 10:52:01.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/21/689421/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-kubernetes","ref":"refs/changes/21/689421/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-kubernetes refs/changes/21/689421/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-kubernetes refs/changes/21/689421/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-kubernetes refs/changes/21/689421/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-kubernetes refs/changes/21/689421/2"}}},"commit":{"parents":[{"commit":"6446f70da0dbc9f69421562c837c487e36fcfc0c","subject":"Merge \"Just check k8s 1.16\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-kubernetes/commit/6446f70da0dbc9f69421562c837c487e36fcfc0c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-10-18 10:42:59.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-10-18 10:51:48.000000000","tz":120},"subject":"Catch right exception for ConnectFailure at activating vif","message":"Catch right exception for ConnectFailure at activating vif\n\nChange-Id: I4122d73ca6e653b11c856fc7d3576d5e8af496fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-kubernetes/commit/d4427604428e3fd69c5728a6fed56359b1889d8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-kubernetes/commit/d4427604428e3fd69c5728a6fed56359b1889d8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
