)]}'
{"id":"openstack%2Fkuryr-kubernetes~832721","triplet_id":"openstack%2Fkuryr-kubernetes~stable%2Fxena~Id966747d8c09ea14bb37880f8fb37447591564b5","project":"openstack/kuryr-kubernetes","branch":"stable/xena","topic":"kubeadm_image_repository-stable/xena","attention_set":{},"removed_from_attention_set":{"28329":{"account":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"last_update":"2022-03-09 14:23:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id966747d8c09ea14bb37880f8fb37447591564b5","subject":"[devstack] Support config image repository for kubeadm","status":"MERGED","created":"2022-03-09 09:19:52.000000000","updated":"2022-03-09 14:24:43.000000000","submitted":"2022-03-09 14:23:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832721-kubeadm_image_repository-stable/xena","cherry_pick_of_change":832168,"cherry_pick_of_patch_set":2,"meta_rev_id":"34e903adbdb6455a8c00d7e21932c53a4b35ab8f","_number":832721,"virtual_id_number":832721,"owner":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-09 14:23:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2022-03-09 09:38:29.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2022-03-09 10:06:35.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-09 10:27:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2022-03-09 11:52:14.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-09 09:38:29.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2022-03-09 10:06:35.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2022-03-09 10:27:53.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-03-09 11:25:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3db9dd849f5a802ab64769911c190e2f7cd3b698","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-03-09 09:19:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"49a5abc4cc20a4ef336a8cfe5872d2cf4bae7b06","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-03-09 09:38:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4810c20d46c41b63471a4ef4a505177bfa3d44e","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-03-09 10:06:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4dfb9ecff2f837ec06d79d3fcfc057cc8de0ef7","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-03-09 10:27:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c1b9bdd62181f9608c8720009bba5ae5db52912c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 11:25:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a87b5cd97164f27a3c910b74779fefe : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f36aa7ce68a4d1abca89e470abf97a0 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f2e4159ef6842f480dab1d84d97d1ee : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1495310bbc6541faa2756d74bb30df35 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f64ac0a98674dbf8724bd29669efbf4 : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b510ba8befe1427da704a85c04e3d0fe : SUCCESS in 7m 47s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/46af720ef8a740ec81e5791298ea875c : SUCCESS in 1h 08m 37s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/56094d9d9e124f7b9fee3721ac119066 : SUCCESS in 1h 30m 51s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/80db504e48ca456d9817d5009efc0c21 : SUCCESS in 1h 44m 10s (non-voting)\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/a5d1706cc99d4f9d8ca51982d821a654 : SUCCESS in 1h 33m 29s\n- kuryr-kubernetes-tempest-network-policy https://zuul.opendev.org/t/openstack/build/527554137b5c44f587346aa05dea7d11 : SUCCESS in 1h 48m 34s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/2162adf9e6b445d18573c86657b558a3 : SUCCESS in 1h 12m 25s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovn https://zuul.opendev.org/t/openstack/build/b22da91ebd884d39acba41b30191f878 : SUCCESS in 1h 17m 10s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/baa9177af82740b8b0e5a69b21321dd2 : SUCCESS in 1h 39m 29s (non-voting)\n- kuryr-kubernetes-tempest-ovn-ipv6 https://zuul.opendev.org/t/openstack/build/19a555f4205e4ff492df45c2784f6eb9 : SUCCESS in 1h 24m 00s\n- kuryr-kubernetes-tempest-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/de7c359e77784aa980d230634e69e65d : SUCCESS in 1h 58m 39s\n- kuryr-kubernetes-e2e-np-ovn-provider-ovn https://zuul.opendev.org/t/openstack/build/da4c89881b30470da3cf174ffdee1ac3 : POST_FAILURE in 55m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0a4cbfb75c515334b834e15b7c886344dd028c2","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-03-09 11:52:14.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"96049ac8e9d5973f149b9903f8a65210fafe79fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 11:58:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b27d54de09eda0905046bcd035182a4ccc5ed5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 14:23:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2422c5cc45ee453986a82118eae3e4ca : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e942c90fba434fbbb9bbd298f167f5b3 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca787ba6c4f54865b64389ea7c159e2c : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/229673d1c6ce4b37bf03da13f8a29e64 : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35e98928789e48979ebdc3b8c98b01bf : SUCCESS in 6m 30s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/b9836eb6e7164cd6a7d047f4c2d77d04 : SUCCESS in 1h 23m 19s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/066a860c5f5e449f8d8d9ab3cfdf6f75 : SUCCESS in 2h 18m 38s\n- kuryr-kubernetes-tempest-ovn https://zuul.opendev.org/t/openstack/build/97f19756b59341e29219bace1bb249e1 : SUCCESS in 1h 43m 13s\n- kuryr-kubernetes-tempest-network-policy https://zuul.opendev.org/t/openstack/build/a306b5d6fb4949a2be27e73b7ea1c56a : SUCCESS in 1h 42m 49s","accounts_in_message":[],"_revision_number":1},{"id":"97534854e402cc7333f332f91e347b01d806b623","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 14:23:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"34e903adbdb6455a8c00d7e21932c53a4b35ab8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 14:24:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e53334593cdb41faace5f7d9bc13e764 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffabb2846cffc34168f11a6816910b951293cf61","revisions":{"ffabb2846cffc34168f11a6816910b951293cf61":{"kind":"REWORK","_number":1,"created":"2022-03-09 09:19:52.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/21/832721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-kubernetes","ref":"refs/changes/21/832721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-kubernetes refs/changes/21/832721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-kubernetes refs/changes/21/832721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-kubernetes refs/changes/21/832721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-kubernetes refs/changes/21/832721/1"}}},"commit":{"parents":[{"commit":"849431fb7cec4738fdf731db29207b51307b4369","subject":"Update Centos version for container images.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-kubernetes/commit/849431fb7cec4738fdf731db29207b51307b4369"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-03-06 08:41:42.000000000","tz":480},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-03-09 09:19:52.000000000","tz":0},"subject":"[devstack] Support config image repository for kubeadm","message":"[devstack] Support config image repository for kubeadm\n\nIn some places of which network environment was limited, kubeadm\ncan\u0027t pull images from k8s.gcr.io. This patch add a variable\n`KURYR_KUBEADMIN_IMAGE_REPOSITORY` in order to the developer who\nlocated in these places can set the kubeadm to pull container images\nfrom repository that they can access.\n\nChange-Id: Id966747d8c09ea14bb37880f8fb37447591564b5\n(cherry picked from commit 8655bb49c80e57be4b773f0a7f8db7c643749b67)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-kubernetes/commit/ffabb2846cffc34168f11a6816910b951293cf61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-kubernetes/commit/ffabb2846cffc34168f11a6816910b951293cf61"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
