)]}'
{"id":"openstack%2Fkuryr-libnetwork~1004367","triplet_id":"openstack%2Fkuryr-libnetwork~master~Ide952578c1531cd8688ada14f907af607c681e5b","project":"openstack/kuryr-libnetwork","branch":"master","attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-09-06 23:58:04.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ide952578c1531cd8688ada14f907af607c681e5b","subject":"Fix typo","status":"NEW","created":"2026-09-06 23:57:13.000000000","updated":"2026-09-08 14:18:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4373de95fa8b6331976b606f728e79a04bcc3324","_number":1004367,"virtual_id_number":1004367,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-07 00:37:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 23:58:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2026-09-07 00:37:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"708b793ee303159427a8089f85d725f17e59f206","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:57:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b30205472214a48377a7ec316bb4e790edface37","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:57:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d5474cb5e4bfd366b2b9226df627407753c093da","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:58:04.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"40c74e2509c6016b66749446b5aa48a445accaf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 00:37:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fe7a9ed34f948c2889128e93a142697\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7a2e6110376446abf8c11d552220fea : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73eca80fe92849efac7715b81a04a5ee : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/743e1ce752314a48a2d5e78d244e6e5e : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/693872d4cb0a4a12b2124ec9be7df9e2 : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/04a5f136eafb46bfb56fd7a58c189f7c : SUCCESS in 5m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f34ffa9440404d62b0cd1ade67717864 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4888abff74e148beb797de5211ef3101 : SUCCESS in 3m 51s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/014ce266071c464ebd05e7ec0f05fe25 : SUCCESS in 12m 14s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/ea55784422744f4d901e10191a4c56aa : SUCCESS in 31m 19s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/8afbb646ec324f84aae0d01dee02c6e9 : FAILURE in 30m 04s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/dbfad49329dc43408c1613c918946722 : SUCCESS in 33m 41s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/12f4a430db554037a24a9213fba66996 : FAILURE in 30m 44s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/d523e13efdf846218d60a4491b9555f5 : SUCCESS in 17m 22s","accounts_in_message":[],"_revision_number":2},{"id":"f6155e78e166b4e8b6dfca0577028efa56d88b02","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 13:29:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4373de95fa8b6331976b606f728e79a04bcc3324","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 14:18:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6eb3772388343aba65b87c438e55bfe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/458667722dd94316ba8a65370b963f68 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d83817fd02204d9daf3a4a07f709bb48 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca277d2be1944201b118f0385bb8c467 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07673d5cb18f4d2098057d7c4e64a2f0 : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d95b626f76b4caeb00a4c8ef3562bda : SUCCESS in 7m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1aebc284facf47adba86b4aff1d413fd : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15ee430b3d4040bfbf764f701f16615b : SUCCESS in 3m 09s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/3f7ee0dc1aad42358d400e6384727d64 : SUCCESS in 28m 57s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/72c35ba01e7a498cae6acb9f916e46fb : SUCCESS in 20m 30s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/00ec01079700488fb7628f0ff9a007b9 : FAILURE in 30m 37s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/c71624af36ce47749729e6a8878adaca : SUCCESS in 16m 26s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/3783e1f124c143cb8528750fde48ff05 : FAILURE in 31m 38s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/49bff94d8b384c989f18137d3af0e2c9 : SUCCESS in 17m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3ab17b06a095edc5b9c95a76462b40157118bf93","revisions":{"39fe3bc8c486827f8235059931238f8124a055a4":{"kind":"REWORK","_number":1,"created":"2026-09-06 23:57:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/67/1004367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/67/1004367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/67/1004367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/67/1004367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/67/1004367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/67/1004367/1"}}},"commit":{"parents":[{"commit":"0e41a69f2c52bf88c312bb3cb35c7f195ff63530","subject":"Merge \"Drop support for Python 3.8, 3.9 and 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/0e41a69f2c52bf88c312bb3cb35c7f195ff63530"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:57:13.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:57:13.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ide952578c1531cd8688ada14f907af607c681e5b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/39fe3bc8c486827f8235059931238f8124a055a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/39fe3bc8c486827f8235059931238f8124a055a4"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"3ab17b06a095edc5b9c95a76462b40157118bf93":{"kind":"REWORK","_number":2,"created":"2026-09-06 23:57:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/67/1004367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/67/1004367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/67/1004367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/67/1004367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/67/1004367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/67/1004367/2"}}},"commit":{"parents":[{"commit":"0e41a69f2c52bf88c312bb3cb35c7f195ff63530","subject":"Merge \"Drop support for Python 3.8, 3.9 and 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/0e41a69f2c52bf88c312bb3cb35c7f195ff63530"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:57:13.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:57:58.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ide952578c1531cd8688ada14f907af607c681e5b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3ab17b06a095edc5b9c95a76462b40157118bf93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3ab17b06a095edc5b9c95a76462b40157118bf93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
