)]}'
{"id":"openstack%2Fkuryr-libnetwork~1004640","triplet_id":"openstack%2Fkuryr-libnetwork~master~I6136c45bc5c7afec5458f656b57031107f5dd49c","project":"openstack/kuryr-libnetwork","branch":"master","attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-09-08 16:18:25.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-08 15:33:20.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I6136c45bc5c7afec5458f656b57031107f5dd49c","subject":"Remove unicode prefix from code","status":"NEW","created":"2026-09-08 13:47:39.000000000","updated":"2026-09-08 16:58:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1306014093cb02d7c389a841bdc4d26023cdc0df","_number":1004640,"virtual_id_number":1004640,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-08 16:58:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 14:36:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2026-09-08 15:25:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de9a3575fa8d43b6f84c9bca1ba9e3a59e11c915","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 13:47:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6e7e9bd21391726a8bad95624922003fcfb4285","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 13:55:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fcc24453247a7f7ce4da758220d8ac1a95fbbead","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 13:57:36.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4a925635db78d1a0e94aed418f2ec28431ccdb0c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 14:09:09.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"67e0a295eb85bc33b260a97f9308fd0165b9026f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 14:35:59.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"e2bb318df235b26a6571a046bed983f6fb5fa57d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 14:36:02.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"db9aed1cab708a656ff6c4a47e45695be1a3396d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 15:25:17.000000000","message":"Patch Set 5: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/915cab90f7be427b9306d7339b6c6276\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/773efb4ea87a4ea1a8917c44a767c9b5 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b385976900f449290c44dce2d3e01a6 : FAILURE in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bbc5000ea12f4ef8bef7c0d658de67a4 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97add1a923a94bd7905f4b074f9ab2a8 : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/431f6dcf434d46d3a24589a7970a5bc8 : SUCCESS in 7m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed6cf8728e07477685890c15026d28ef : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/155931d4062d464fa28963075937cd44 : SUCCESS in 4m 06s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/2d0c994605e4445c83b6320dd93ed800 : SUCCESS in 30m 45s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/21cc2683c9ec4ec391623e50a0eca2eb : SUCCESS in 34m 29s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/0fc36a6ecae0492097bdd7ff040fed5f : FAILURE in 31m 45s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/684637242d7647e898f31df583a575e3 : SUCCESS in 30m 51s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/8437d6eaf44e460c8cc0ea10ee94735c : FAILURE in 31m 05s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/7ca544728ee64486884ee78d0ee7f73e : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":5},{"id":"16b0dace2e0d5c87a8fca3014b38b9e58c1601e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 15:29:48.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"354583c3c592d0f8fba6fe3a43c422dcde1f87ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 15:33:09.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"db3a32fbfa8f88c9eccbda9c951b242ed2bc2a04","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 15:33:20.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":7},{"id":"89fe547478b1aa00f47bc7b7963543fa4de7d903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:09:25.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d90dc9741500472aa961889e6bfcaf9a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2e3a206bbb7423bba84ae10b51215fc : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aa05fb45e8a4866b5fa25599725e0d5 : FAILURE in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/093d2862f71c467282c89f89b36e15c6 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c2df988aa33496cb78a069901ae13ff : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e4f702ac4d514ac0ad92b2a14547792e : SUCCESS in 6m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c330a5a8daf4822b30a63289fc6ff05 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd38a69cbc14b189babe85f40d6bc65 : SUCCESS in 3m 48s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/e87b3412b9b146cea6b4cdad8293b9cc : SUCCESS in 29m 54s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/f01affba88764d10b8398ff842c308c5 : SUCCESS in 31m 45s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/1c2118ea85e249e2bebb2b0679ca190e : FAILURE in 30m 35s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/3e3526595eb74b2bbf8004ee1c4fd762 : SUCCESS in 33m 40s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/6b2a1c1eb0fb453282f315759d769778 : FAILURE in 16m 09s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/c0d163a65e2a444198886b64f2459f48 : SUCCESS in 19m 59s","accounts_in_message":[],"_revision_number":7},{"id":"533b35ab066e2f470297d2090c3978133f86ca41","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 16:15:36.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"4a88ef5ea3e952db9b79d4ef4cc5b87ed4190359","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 16:18:12.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"0f307a47f260dc6f9245e0b308a614fb6ee4a237","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-08 16:18:25.000000000","message":"Patch Set 9:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":9},{"id":"1306014093cb02d7c389a841bdc4d26023cdc0df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:58:33.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51268f3393e047909ea8f9fdc03a2656\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0799652d0dbf4cb5a39c26f8a1529583 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7526f1a9ef6a463b961bc91a398fd322 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa2c886a91f64983b155c5b909053118 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78214322deae465180b318edf1fb04c7 : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5961391c6c5545b29410ea43a89e168f : SUCCESS in 5m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3cd90fdaa864a46b7f7ed9750e8ecaa : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd168edd9a934b6c9b62df25f5b072ed : SUCCESS in 2m 43s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/cc8ea6de4f91496a9fd3830f78e0f2f2 : SUCCESS in 30m 56s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/830f1ee8349c4d66bd3a0ec495f097b4 : SUCCESS in 32m 31s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/28183406307f46b4b7c5d99062c42bfe : FAILURE in 29m 53s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/3d017ce365204f1d86a721a38d232bb6 : SUCCESS in 27m 57s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/cc1e7719f44346be8462003865dcf54a : FAILURE in 30m 15s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/05fa09ccf926486c9c5792da18a0ab9c : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"f9d1d8021ace203179c779f70bad11692fcf66db","revisions":{"b84a8443e3be1e7e45d686973ee154092cb2e406":{"kind":"REWORK","_number":1,"created":"2026-09-08 13:47:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/40/1004640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/40/1004640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/1"}}},"commit":{"parents":[{"commit":"0e41a69f2c52bf88c312bb3cb35c7f195ff63530","subject":"Merge \"Drop support for Python 3.8, 3.9 and 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/0e41a69f2c52bf88c312bb3cb35c7f195ff63530"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:47:39.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:47:39.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I6136c45bc5c7afec5458f656b57031107f5dd49c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/b84a8443e3be1e7e45d686973ee154092cb2e406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/b84a8443e3be1e7e45d686973ee154092cb2e406"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"60c9be8c314950c5c439b74d921b771f23b9dc54":{"kind":"REWORK","_number":2,"created":"2026-09-08 13:55:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/40/1004640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/40/1004640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/2"}}},"commit":{"parents":[{"commit":"0e41a69f2c52bf88c312bb3cb35c7f195ff63530","subject":"Merge \"Drop support for Python 3.8, 3.9 and 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/0e41a69f2c52bf88c312bb3cb35c7f195ff63530"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:47:39.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:55:07.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I6136c45bc5c7afec5458f656b57031107f5dd49c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/60c9be8c314950c5c439b74d921b771f23b9dc54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/60c9be8c314950c5c439b74d921b771f23b9dc54"}]},"branch":"refs/heads/master"},"5e8efe9b90f20a6d554da2c7e7b8e465d31e4e5a":{"kind":"REWORK","_number":3,"created":"2026-09-08 13:57:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/40/1004640/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/40/1004640/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/3"}}},"commit":{"parents":[{"commit":"0e41a69f2c52bf88c312bb3cb35c7f195ff63530","subject":"Merge \"Drop support for Python 3.8, 3.9 and 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/0e41a69f2c52bf88c312bb3cb35c7f195ff63530"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:47:39.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:57:36.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I6136c45bc5c7afec5458f656b57031107f5dd49c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/5e8efe9b90f20a6d554da2c7e7b8e465d31e4e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/5e8efe9b90f20a6d554da2c7e7b8e465d31e4e5a"}]},"branch":"refs/heads/master"},"7140d438140d23e5d11158f06769b9cd82833a1a":{"kind":"REWORK","_number":4,"created":"2026-09-08 14:09:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/40/1004640/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/40/1004640/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/4"}}},"commit":{"parents":[{"commit":"0e41a69f2c52bf88c312bb3cb35c7f195ff63530","subject":"Merge \"Drop support for Python 3.8, 3.9 and 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/0e41a69f2c52bf88c312bb3cb35c7f195ff63530"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:47:39.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 14:09:09.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I6136c45bc5c7afec5458f656b57031107f5dd49c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/7140d438140d23e5d11158f06769b9cd82833a1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/7140d438140d23e5d11158f06769b9cd82833a1a"}]},"branch":"refs/heads/master"},"119151d05a07be7b2aaed2f0be9e4f096c0261ee":{"kind":"REWORK","_number":5,"created":"2026-09-08 14:35:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/40/1004640/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/40/1004640/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/5"}}},"commit":{"parents":[{"commit":"0e41a69f2c52bf88c312bb3cb35c7f195ff63530","subject":"Merge \"Drop support for Python 3.8, 3.9 and 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/0e41a69f2c52bf88c312bb3cb35c7f195ff63530"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:47:39.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 14:35:59.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I6136c45bc5c7afec5458f656b57031107f5dd49c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/119151d05a07be7b2aaed2f0be9e4f096c0261ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/119151d05a07be7b2aaed2f0be9e4f096c0261ee"}]},"branch":"refs/heads/master"},"048a53bc124a8a38d38dc89b8c0d8d1f3e7bfd72":{"kind":"REWORK","_number":6,"created":"2026-09-08 15:29:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/40/1004640/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/40/1004640/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/6"}}},"commit":{"parents":[{"commit":"0e41a69f2c52bf88c312bb3cb35c7f195ff63530","subject":"Merge \"Drop support for Python 3.8, 3.9 and 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/0e41a69f2c52bf88c312bb3cb35c7f195ff63530"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:47:39.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 15:29:48.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I6136c45bc5c7afec5458f656b57031107f5dd49c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/048a53bc124a8a38d38dc89b8c0d8d1f3e7bfd72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/048a53bc124a8a38d38dc89b8c0d8d1f3e7bfd72"}]},"branch":"refs/heads/master"},"1294bcd2c70786584d96ab7c0e975ad72ee5b4b9":{"kind":"REWORK","_number":7,"created":"2026-09-08 15:33:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/40/1004640/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/40/1004640/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/7"}}},"commit":{"parents":[{"commit":"0e41a69f2c52bf88c312bb3cb35c7f195ff63530","subject":"Merge \"Drop support for Python 3.8, 3.9 and 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/0e41a69f2c52bf88c312bb3cb35c7f195ff63530"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:47:39.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 15:33:09.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I6136c45bc5c7afec5458f656b57031107f5dd49c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/1294bcd2c70786584d96ab7c0e975ad72ee5b4b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/1294bcd2c70786584d96ab7c0e975ad72ee5b4b9"}]},"branch":"refs/heads/master"},"5a60a49b02f07ccbb7e76f0e796e4849a08dfff0":{"kind":"REWORK","_number":8,"created":"2026-09-08 16:15:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/40/1004640/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/40/1004640/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/8"}}},"commit":{"parents":[{"commit":"0e41a69f2c52bf88c312bb3cb35c7f195ff63530","subject":"Merge \"Drop support for Python 3.8, 3.9 and 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/0e41a69f2c52bf88c312bb3cb35c7f195ff63530"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:47:39.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 16:15:36.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I6136c45bc5c7afec5458f656b57031107f5dd49c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/5a60a49b02f07ccbb7e76f0e796e4849a08dfff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/5a60a49b02f07ccbb7e76f0e796e4849a08dfff0"}]},"branch":"refs/heads/master"},"f9d1d8021ace203179c779f70bad11692fcf66db":{"kind":"REWORK","_number":9,"created":"2026-09-08 16:18:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/40/1004640/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/40/1004640/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/1004640/9"}}},"commit":{"parents":[{"commit":"0e41a69f2c52bf88c312bb3cb35c7f195ff63530","subject":"Merge \"Drop support for Python 3.8, 3.9 and 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/0e41a69f2c52bf88c312bb3cb35c7f195ff63530"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 13:47:39.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-08 16:18:12.000000000","tz":0},"subject":"Remove unicode prefix from code","message":"Remove unicode prefix from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I6136c45bc5c7afec5458f656b57031107f5dd49c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/f9d1d8021ace203179c779f70bad11692fcf66db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/f9d1d8021ace203179c779f70bad11692fcf66db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
