)]}'
{"id":"openstack%2Fkuryr-libnetwork~656414","triplet_id":"openstack%2Fkuryr-libnetwork~master~I4fee6ebcca6cbb1c610fa19c4294ef5a45b50709","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"I4fee6ebcca6cbb1c610fa19c4294ef5a45b50709","subject":"Replace port 35357 with port 5000","status":"MERGED","created":"2019-04-29 22:17:04.000000000","updated":"2019-05-08 10:30:26.000000000","submitted":"2019-05-08 10:30:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656414-1557311426265-a31b8f89","meta_rev_id":"d5846f58d560eb8bc16e78dff8fc9d7314e1687b","_number":656414,"virtual_id_number":656414,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-29 22:17:30.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":2,"date":"2019-05-08 10:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":2,"date":"2019-05-08 09:53:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":2,"date":"2019-04-30 19:04:34.000000000","_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"date":"2019-04-29 22:25:40.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":1,"date":"2019-05-08 09:53:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"date":"2019-04-29 22:25:40.000000000","_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-29 22:25:40.000000000","_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-29 22:25:40.000000000","updated_by":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"reviewer":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"state":"REVIEWER"},{"updated":"2019-04-29 22:25:40.000000000","updated_by":{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},"reviewer":{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},"state":"REVIEWER"},{"updated":"2019-04-29 22:25:40.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2019-04-30 19:04:34.000000000","updated_by":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"reviewer":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"state":"REVIEWER"},{"updated":"2019-05-08 09:53:52.000000000","updated_by":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"reviewer":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"state":"REVIEWER"},{"updated":"2019-05-08 10:30:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95382b4de3faa600da9a82a657fcad092a85071b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-29 22:17:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba0f140ad6d2cfc8cbe81d993c30ca0c803eb848","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-29 22:17:30.000000000","message":"Patch Set 1:\n\nNote: need to backport all the way to Queens","accounts_in_message":[],"_revision_number":1},{"id":"59609661b0833bc343df1c4a203c5d49c3702add","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 23:06:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/14/656414/1/check/test-release-openstack/e892240/ : SUCCESS in 3m 28s\n- openstack-tox-cover http://logs.openstack.org/14/656414/1/check/openstack-tox-cover/5b55743/cover/ : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/656414/1/check/openstack-tox-lower-constraints/262a576/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 http://logs.openstack.org/14/656414/1/check/openstack-tox-pep8/6c6a8f1/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/14/656414/1/check/openstack-tox-py27/cb48c8e/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/14/656414/1/check/openstack-tox-py36/0e08505/ : SUCCESS in 5m 41s\n- openstack-tox-py37 http://logs.openstack.org/14/656414/1/check/openstack-tox-py37/5302cf6/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/14/656414/1/check/openstack-tox-docs/2f8dfa5/html/ : SUCCESS in 6m 37s\n- kuryr-libnetwork-install http://logs.openstack.org/14/656414/1/check/kuryr-libnetwork-install/88453fd/ : SUCCESS in 28m 49s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack http://logs.openstack.org/14/656414/1/check/kuryr-libnetwork-kuryr_lib-src-fullstack/830c915/ : SUCCESS in 40m 29s (non-voting)\n- kuryr-libnetwork-fullstack http://logs.openstack.org/14/656414/1/check/kuryr-libnetwork-fullstack/a4537dc/ : SUCCESS in 30m 58s\n- kuryr-libnetwork-kuryr_lib-src-rally http://logs.openstack.org/14/656414/1/check/kuryr-libnetwork-kuryr_lib-src-rally/98f95a4/ : SUCCESS in 40m 12s (non-voting)\n- kuryr-libnetwork-fullstack-py3 http://logs.openstack.org/14/656414/1/check/kuryr-libnetwork-fullstack-py3/fb8eb19/ : SUCCESS in 37m 08s\n- kuryr-libnetwork-rally http://logs.openstack.org/14/656414/1/check/kuryr-libnetwork-rally/5230482/ : SUCCESS in 44m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43f0200207fe19ab94c8ce61d09f47d39b4bd11b","author":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"date":"2019-04-30 19:04:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"309fade932cd953e56f1318ee58b15572de7a5a0","author":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"date":"2019-05-08 09:53:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f3cd29747655f948f20332f7d86c9a3b3d47675","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 09:54:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e044f88f70318c2457d87ababa244b77817dd6fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 10:30:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/14/656414/1/gate/test-release-openstack/6ece074/ : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/656414/1/gate/openstack-tox-lower-constraints/7706ce1/ : SUCCESS in 4m 26s\n- openstack-tox-pep8 http://logs.openstack.org/14/656414/1/gate/openstack-tox-pep8/acc8739/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/14/656414/1/gate/openstack-tox-py27/5033f90/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/14/656414/1/gate/openstack-tox-py36/f9668fb/ : SUCCESS in 4m 54s\n- openstack-tox-py37 http://logs.openstack.org/14/656414/1/gate/openstack-tox-py37/e7a78bd/ : SUCCESS in 6m 23s\n- openstack-tox-docs http://logs.openstack.org/14/656414/1/gate/openstack-tox-docs/0fcf310/html/ : SUCCESS in 5m 51s\n- kuryr-libnetwork-kuryr_lib-src-fullstack http://logs.openstack.org/14/656414/1/gate/kuryr-libnetwork-kuryr_lib-src-fullstack/65cc587/ : SUCCESS in 28m 10s (non-voting)\n- kuryr-libnetwork-fullstack http://logs.openstack.org/14/656414/1/gate/kuryr-libnetwork-fullstack/97b21a6/ : SUCCESS in 30m 15s\n- kuryr-libnetwork-fullstack-py3 http://logs.openstack.org/14/656414/1/gate/kuryr-libnetwork-fullstack-py3/f058825/ : SUCCESS in 35m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d5846f58d560eb8bc16e78dff8fc9d7314e1687b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 10:30:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c528e9be2fe244329c83429d5375da5ab398561f","revisions":{"c528e9be2fe244329c83429d5375da5ab398561f":{"kind":"REWORK","_number":1,"created":"2019-04-29 22:17:04.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/14/656414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/14/656414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/14/656414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/14/656414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/14/656414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/14/656414/1"}}},"commit":{"parents":[{"commit":"9665dafdd1d0f112331c5e9526b72a883248625d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/9665dafdd1d0f112331c5e9526b72a883248625d"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-29 22:15:17.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-29 22:17:03.000000000","tz":0},"subject":"Replace port 35357 with port 5000","message":"Replace port 35357 with port 5000\n\nStarting from Keystone v3, the port 35357 is not used.\nSee: https://review.opendev.org/#/c/541857/\n\nChange-Id: I4fee6ebcca6cbb1c610fa19c4294ef5a45b50709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/c528e9be2fe244329c83429d5375da5ab398561f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/c528e9be2fe244329c83429d5375da5ab398561f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
