)]}'
{"id":"openstack%2Fkuryr-libnetwork~672895","triplet_id":"openstack%2Fkuryr-libnetwork~master~Ibed8ffe0effaf523fff2e5eaebcc0dad45c62d54","project":"openstack/kuryr-libnetwork","branch":"master","topic":"openstackdocstheme","hashtags":[],"change_id":"Ibed8ffe0effaf523fff2e5eaebcc0dad45c62d54","subject":"Bump the openstackdocstheme extension to 1.20","status":"NEW","created":"2019-07-26 07:20:51.000000000","updated":"2019-09-16 05:27:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fbfc595a75d34d45b05db5aa468c3a75a9c7ea54","_number":672895,"virtual_id_number":672895,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":-1,"date":"2019-08-01 03:24:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":-1,"date":"2019-09-16 05:27:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-27 21:16:24.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2019-08-01 03:24:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-16 05:27:14.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"}],"messages":[{"id":"de2053e3b99710816acffa52f5d53df98f351b70","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-26 07:20:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0994f7c09743c7dc8fcf68d2952d1cbfd445cbf6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 08:16:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/95/672895/1/check/openstack-tox-cover/9f00daa/cover/ : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/672895/1/check/openstack-tox-lower-constraints/938fcb3/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/95/672895/1/check/openstack-tox-pep8/3c18cc6/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/95/672895/1/check/openstack-tox-py27/36c4dda/ : SUCCESS in 7m 37s\n- openstack-tox-py36 http://logs.openstack.org/95/672895/1/check/openstack-tox-py36/23676ac/ : SUCCESS in 4m 45s\n- openstack-tox-py37 http://logs.openstack.org/95/672895/1/check/openstack-tox-py37/c4e7474/ : SUCCESS in 3m 48s\n- build-openstack-releasenotes http://logs.openstack.org/95/672895/1/check/build-openstack-releasenotes/3e01bc6/html/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/95/672895/1/check/openstack-tox-docs/73d17c3/html/ : SUCCESS in 5m 31s\n- requirements-check http://logs.openstack.org/95/672895/1/check/requirements-check/77288ad/ : FAILURE in 4m 15s\n- kuryr-libnetwork-install http://logs.openstack.org/95/672895/1/check/kuryr-libnetwork-install/6bb1f73/ : SUCCESS in 32m 15s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack http://logs.openstack.org/95/672895/1/check/kuryr-libnetwork-kuryr_lib-src-fullstack/44ec707/ : SUCCESS in 41m 06s (non-voting)\n- kuryr-libnetwork-fullstack http://logs.openstack.org/95/672895/1/check/kuryr-libnetwork-fullstack/ac29442/ : SUCCESS in 35m 04s\n- kuryr-libnetwork-kuryr_lib-src-rally http://logs.openstack.org/95/672895/1/check/kuryr-libnetwork-kuryr_lib-src-rally/1132e81/ : SUCCESS in 49m 29s (non-voting)\n- kuryr-libnetwork-fullstack-py3 http://logs.openstack.org/95/672895/1/check/kuryr-libnetwork-fullstack-py3/d1ab3f9/ : SUCCESS in 29m 08s\n- kuryr-libnetwork-rally http://logs.openstack.org/95/672895/1/check/kuryr-libnetwork-rally/226063d/ : SUCCESS in 45m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eccb506f12340a6a1f6a9138883df63bdf3c3714","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-07-27 21:16:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"249a495f855f61664f140279ce479a48454a5f47","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-01 02:36:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e26c39253555a101bff308516a68638d76183e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 03:24:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/95/672895/2/check/openstack-tox-cover/12466be/cover/ : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints https://logs.opendev.org/95/672895/2/check/openstack-tox-lower-constraints/c99395d/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://logs.opendev.org/95/672895/2/check/openstack-tox-pep8/76b8455/ : SUCCESS in 4m 09s\n- openstack-tox-py27 https://logs.opendev.org/95/672895/2/check/openstack-tox-py27/56fab91/ : SUCCESS in 4m 47s\n- openstack-tox-py36 https://logs.opendev.org/95/672895/2/check/openstack-tox-py36/64373b0/ : SUCCESS in 4m 20s\n- openstack-tox-py37 https://logs.opendev.org/95/672895/2/check/openstack-tox-py37/d6eb6e4/ : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://logs.opendev.org/95/672895/2/check/build-openstack-releasenotes/82ffbf0/html/ : SUCCESS in 3m 35s\n- openstack-tox-docs https://logs.opendev.org/95/672895/2/check/openstack-tox-docs/066c5b1/html/ : SUCCESS in 3m 56s\n- requirements-check https://logs.opendev.org/95/672895/2/check/requirements-check/e29baf1/ : FAILURE in 2m 18s\n- kuryr-libnetwork-install https://logs.opendev.org/95/672895/2/check/kuryr-libnetwork-install/9535cdf/ : SUCCESS in 32m 51s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://logs.opendev.org/95/672895/2/check/kuryr-libnetwork-kuryr_lib-src-fullstack/c8ba2fb/ : SUCCESS in 31m 30s (non-voting)\n- kuryr-libnetwork-fullstack https://logs.opendev.org/95/672895/2/check/kuryr-libnetwork-fullstack/32f1009/ : SUCCESS in 33m 27s\n- kuryr-libnetwork-kuryr_lib-src-rally https://logs.opendev.org/95/672895/2/check/kuryr-libnetwork-kuryr_lib-src-rally/ad81c81/ : SUCCESS in 43m 11s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://logs.opendev.org/95/672895/2/check/kuryr-libnetwork-fullstack-py3/c22cc35/ : SUCCESS in 34m 21s\n- kuryr-libnetwork-rally https://logs.opendev.org/95/672895/2/check/kuryr-libnetwork-rally/6be30f3/ : SUCCESS in 43m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b49086a6e110d5efb038301a0e61e012af66984b","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-16 05:27:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"493ab79317af469e5b793043bedbed136d5ff292","revisions":{"8c6f96f292d5b8511133a215e4c9bad841ef18e0":{"kind":"REWORK","_number":1,"created":"2019-07-26 07:20:51.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/95/672895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/95/672895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/95/672895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/95/672895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/95/672895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/95/672895/1"}}},"commit":{"parents":[{"commit":"5264dee9649399114c436eca8ac27f5d01905273","subject":"Merge \"Update install guide for master branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/5264dee9649399114c436eca8ac27f5d01905273"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-26 07:20:09.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-26 07:21:02.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Ibed8ffe0effaf523fff2e5eaebcc0dad45c62d54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8c6f96f292d5b8511133a215e4c9bad841ef18e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8c6f96f292d5b8511133a215e4c9bad841ef18e0"}]},"branch":"refs/heads/master"},"493ab79317af469e5b793043bedbed136d5ff292":{"kind":"REWORK","_number":2,"created":"2019-08-01 02:36:47.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/95/672895/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/95/672895/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/95/672895/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/95/672895/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/95/672895/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/95/672895/2"}}},"commit":{"parents":[{"commit":"5264dee9649399114c436eca8ac27f5d01905273","subject":"Merge \"Update install guide for master branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/5264dee9649399114c436eca8ac27f5d01905273"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-26 07:20:09.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-01 02:36:55.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Ibed8ffe0effaf523fff2e5eaebcc0dad45c62d54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/493ab79317af469e5b793043bedbed136d5ff292"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/493ab79317af469e5b793043bedbed136d5ff292"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
