)]}'
{"id":"openstack%2Fkuryr-libnetwork~677593","triplet_id":"openstack%2Fkuryr-libnetwork~master~I1450759319d82d3533d887e45d01b4185256786a","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"I1450759319d82d3533d887e45d01b4185256786a","subject":"Fix ipv6 issue on devstack","status":"MERGED","created":"2019-08-21 02:05:28.000000000","updated":"2019-08-22 11:55:44.000000000","submitted":"2019-08-22 11:54:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677593-1566474862988-37b48900","meta_rev_id":"a3ea7914c2198f196fbdeffb94fd38c2a5ae9100","_number":677593,"virtual_id_number":677593,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-22 02:34:02.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":2,"date":"2019-08-22 11:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-08-22 10:56:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2019-08-22 10:56:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"date":"2019-08-22 04:07:55.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"date":"2019-08-22 04:07:55.000000000","_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-22 04:07:55.000000000","_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"date":"2019-08-22 04:07:55.000000000","_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 04:07:55.000000000","updated_by":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"reviewer":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"state":"REVIEWER"},{"updated":"2019-08-22 04:07:55.000000000","updated_by":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"reviewer":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"state":"REVIEWER"},{"updated":"2019-08-22 04:07:55.000000000","updated_by":{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},"reviewer":{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},"state":"REVIEWER"},{"updated":"2019-08-22 04:07:55.000000000","updated_by":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"reviewer":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"state":"REVIEWER"},{"updated":"2019-08-22 10:56:41.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2019-08-22 11:54:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c06e44f0f9b2b6c8b2790a2955ff1b1d0b2d5e13","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-21 02:05:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bfdcd1dd7400bfd8098c3f7f60f24cb1820dc0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 02:58:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/342f15871a64449f9a13089ca258c803 : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99cc9123fd8549d7925f5055e94aee39 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d943ff48114295a8e15de1a40c3382 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88d7c7ccefcf4a57892fbdfc21f09442 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8825b610e97549e38552e83d44ada962 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c8e13fee4044274b88e1816b8640f3b : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc2e69d9d2c3488c969fc0bbc3377c9d : SUCCESS in 4m 58s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/188333ef47f94a42bfd8935b8651ae03 : SUCCESS in 26m 52s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/23c2fef117764f3aa13bbbfdf7f720e0 : SUCCESS in 34m 57s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/c7fffe21220c457c9d2c8bd229324e49 : SUCCESS in 37m 44s\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/747a052133da4cd7a39da7ae94c5ca14 : SUCCESS in 47m 01s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/5c908be4eccf4a33b9da7c3861441039 : SUCCESS in 50m 55s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/802d9f3cd4d043afa56902c6c59c0112 : SUCCESS in 45m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f5967382e6b81b25c6f6a56be6634e992e55502","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-22 01:07:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"772aa19cca2a39a15d08b8d25dd2e4926fcb98bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 02:03:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd4932b3267d48e388fa4716aa70eca5 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9282452bdf624b3d9b5c69843ae7fa1a : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d13ce515c66448e9f8f0b584c007407 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/539f347784014565b0c87e45508414a8 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51583575d8264b5d91322511becd6814 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ef00b4d491c4765805f433d5ad0b720 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e45afb0a69491aaa8173f675343af1 : SUCCESS in 3m 48s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/999463ec48e341c39f61c473ad20776d : SUCCESS in 41m 19s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/05e2502a6ec048b6af8aeeb54de87ef9 : SUCCESS in 33m 51s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/dd80194556714750b1449d29f5e699fd : SUCCESS in 34m 31s\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/b7337448dd834d19aaf7ac6462e5fae0 : SUCCESS in 54m 17s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/630ee093a3e842c985b7e941244f97c9 : FAILURE in 32m 04s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/eba7f3833e9d40d781336e4bfd50ca2b : SUCCESS in 49m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f6f96f090ecf36fd29614109fbcc17819dba24b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-22 02:34:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"33fb2094f0d3e915051dddc70944596d67378dff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 03:29:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f5e3178a79146e4a9062d5407a1a65c : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33f223d6a98f4e5fb517aa1819a3039a : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/747885ee8b4644ae8d947c9c82650ba3 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba37385aa998457f9bb4e298f13e52da : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f77bf699fe1549638d9beb811631027f : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6b1841b41e4414293091073c5878def : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19c3d8eb9fe0472089f2f0c9d4fc15ed : SUCCESS in 4m 35s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/fb2c259ac5864138bc417bb984d2144b : SUCCESS in 31m 21s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/00743ec92bc54a5d9ff0c6c28b70a56d : SUCCESS in 34m 47s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/48c9374a63374536a54f9491e2e7d29f : SUCCESS in 37m 04s\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/fd60e568b42d484a93bdeda6c9485c72 : SUCCESS in 47m 27s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/c41f28fe8c134fb6b3f3ad1f3ec381aa : SUCCESS in 33m 23s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/221ff3406d604d6dbe33d81a4ddd6214 : SUCCESS in 47m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae3d5c93a04dc3046bae23b08e523a6a05ffa62f","author":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"date":"2019-08-22 10:56:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e09ebccb1ea980a17bf51024e2711e8e06e67fcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:56:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f063248acb5f506f392c5be805ccae58e3d9b79b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:54:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d04f63e05fd431586ecde5202dbff91 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d2eb6f09f94d0bad0a5111cde57855 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/731ea2cba5f34f12b7727e6b3a385a0c : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ab3f4c5bd8941a7b491aacdeb58b654 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1dbd335d0a04fa58c6f3dd2d1a20f58 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167fec512b5e40f1881cf9f2b9f92439 : SUCCESS in 5m 16s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/979949b9fdc34428ba0f07bf95c7bbad : SUCCESS in 35m 24s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/b38fb16fe4ef45f094712f0b97361a74 : SUCCESS in 33m 32s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/fa3e004fea70497cae73d2401206c599 : SUCCESS in 33m 20s","accounts_in_message":[],"_revision_number":2},{"id":"9b7f80904816e126b98fcb82d6fcc71784fe6762","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:54:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a3ea7914c2198f196fbdeffb94fd38c2a5ae9100","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:55:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9001fcfef8042d7bd8fdabb959938bf : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b63cd9ae4dd9fbf6ea95adbc4a0006534cd7e7a9","revisions":{"6fffb856412c0101a3beb39fe95aa324ff7d9385":{"kind":"REWORK","_number":1,"created":"2019-08-21 02:05:28.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/93/677593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/93/677593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/93/677593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/93/677593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/93/677593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/93/677593/1"}}},"commit":{"parents":[{"commit":"5264dee9649399114c436eca8ac27f5d01905273","subject":"Merge \"Update install guide for master branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/5264dee9649399114c436eca8ac27f5d01905273"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-08-21 02:03:06.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-08-21 02:05:28.000000000","tz":0},"subject":"Fix ipv6 issue on devstack","message":"Fix ipv6 issue on devstack\n\nChange-Id: I1450759319d82d3533d887e45d01b4185256786a\nStory: #2005477\nTask: #35942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/6fffb856412c0101a3beb39fe95aa324ff7d9385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/6fffb856412c0101a3beb39fe95aa324ff7d9385"}]},"branch":"refs/heads/master"},"b63cd9ae4dd9fbf6ea95adbc4a0006534cd7e7a9":{"kind":"REWORK","_number":2,"created":"2019-08-22 01:07:12.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/93/677593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/93/677593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/93/677593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/93/677593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/93/677593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/93/677593/2"}}},"commit":{"parents":[{"commit":"5264dee9649399114c436eca8ac27f5d01905273","subject":"Merge \"Update install guide for master branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/5264dee9649399114c436eca8ac27f5d01905273"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-08-21 02:03:06.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-08-22 01:07:06.000000000","tz":0},"subject":"Fix ipv6 issue on devstack","message":"Fix ipv6 issue on devstack\n\nChange-Id: I1450759319d82d3533d887e45d01b4185256786a\nStory: #2005477\nTask: #35942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/b63cd9ae4dd9fbf6ea95adbc4a0006534cd7e7a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/b63cd9ae4dd9fbf6ea95adbc4a0006534cd7e7a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
