)]}'
{"id":"openstack%2Fkuryr-libnetwork~683894","triplet_id":"openstack%2Fkuryr-libnetwork~stable%2Ftrain~Icd3e100a4254a0f8b77e55290b8de912787c84a0","project":"openstack/kuryr-libnetwork","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"Icd3e100a4254a0f8b77e55290b8de912787c84a0","subject":"Update .gitreview for stable/train","status":"MERGED","created":"2019-09-23 08:58:12.000000000","updated":"2019-10-03 03:32:55.000000000","submitted":"2019-10-03 03:31:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683894-1570073480836-fe29816b","meta_rev_id":"a4a310c80eed5e7d9c0979c9894a49359b848ebf","_number":683894,"virtual_id_number":683894,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 03:31:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-10-03 02:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-10-03 02:51:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 02:51:06.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2019-10-03 03:31:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4924fa3fe31ea93238376a28bcc77eef0abd1551","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-09-23 08:58:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5418885225ba23e54b2c7e07ac184f885ee8a607","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 09:49:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0c9c30ec4454dedbc405cf5c895b943 : FAILURE in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8745ff7d48a140769224505ce6b4fd33 : FAILURE in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afefaae06f104e33b7d0dcf275868adf : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd8b3a026aad45f5a4f17b82077a1671 : FAILURE in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bdc2d6ce2694aafa0c82e1f06c7ea51 : FAILURE in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1565c0e0f0b4fc2836a102add568b49 : FAILURE in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ded6cfa1cc014d3a9969a17912a294a9 : SUCCESS in 4m 00s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/544a24960d884911a81bb87a2d831793 : SUCCESS in 29m 28s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/9e1a84d9da6a49e99986779010a13a7d : SUCCESS in 34m 37s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/84af940f7aad45a581c8944dd35b0ea0 : SUCCESS in 35m 57s","accounts_in_message":[],"_revision_number":1},{"id":"906e8b3d8e874c7564bd59fec18d840450dfb96c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-01 14:24:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a0d03347872a25f32d438d2188d0074c34f284a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 15:11:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a9a5857f68040f9a5426179e4e80752 : FAILURE in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dd456313d224b72a19dd669ac90f506 : FAILURE in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4a3f59b7c654ff3a6c627f47e12bb4f : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4b475e4f9af4557a46e9e97568f6ac4 : FAILURE in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/728b0df1682a4245acd78863699e0650 : FAILURE in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8e7f33586fa4064998048c05d0d14f1 : FAILURE in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d1f5efc1a2415fb20c2c7ac8453dcb : SUCCESS in 4m 36s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/b64529623e4c4fe1a787099b6fe27bc6 : SUCCESS in 33m 17s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/83a0bbf55de64a5286b134ad8674b991 : SUCCESS in 39m 22s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/e208317a16ea417b92a91bee83d73149 : FAILURE in 32m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f801bf1552ea5d88fed18f16d8114d874fa57fee","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-03 02:01:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f7d822547f361cf5a18a928207a62cd764ef895f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 02:44:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1355b9ee41ad4e8394bb87661fd7e6fe : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ed9f91f67124eb9a733cffeab29adaf : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/385b456412904873b61f6f38efc2789a : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45dc6495317c4042bdd02eb167ccd123 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1167601bdb544512972759a9ac3ec64b : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9f02536793a4fe9b98fa6388bd6e7d1 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03568401f46d416e96e09481fd239e83 : SUCCESS in 3m 34s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/89d4e6510f64441e9214460c2592cd05 : SUCCESS in 33m 52s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/7d0ade82ca5543feac2e00038dfb02f4 : SUCCESS in 36m 57s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/8bbc72dfa540438ea59a75f6b32525f0 : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ba4f8037a0d18d461afa8f27cb7ce97ad0e0334c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-03 02:51:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae42c6fa84b33ebe743ec59b1608b0d9f79e4e03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 02:51:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7df79e6914779e66b99f7511476dc7ccefbcfeaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 03:31:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f6512ea9688459bb97a49f91df233da : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/031e5c9a757242ff95678f13e0e11aff : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8033e08323514f9fba2716bf1363f20b : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bab5f1b97ea415bb8be3ccb87de576a : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4017c4a0ecf429e84c5aa9740f02655 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce8a8d22bdb4d6fa9a41849e2dbafb3 : SUCCESS in 4m 33s\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/c2f55200c50244d99040900778ccf3e0 : SUCCESS in 33m 09s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/ef130d22c890442abaaa2a56bd62b71f : SUCCESS in 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3d18c49876ff6984a292a292269b61a824975da1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 03:31:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"52104e374f53e8ee136e489547155335ee34f7f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 03:32:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe5944b39904cb68e898ccd32e06d83 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71ed2d2a2bcddacea7d3481ce2cefea3f2d5f6eb","revisions":{"71ed2d2a2bcddacea7d3481ce2cefea3f2d5f6eb":{"kind":"REWORK","_number":1,"created":"2019-09-23 08:58:12.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/94/683894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/94/683894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/94/683894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/94/683894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/94/683894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/94/683894/1"}}},"commit":{"parents":[{"commit":"622ded3e28c37616a667b2901a3fa2d2ce277de6","subject":"Merge \"Fix ipv6 issue on devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/622ded3e28c37616a667b2901a3fa2d2ce277de6"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-23 08:58:10.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-23 08:58:10.000000000","tz":0},"subject":"Update .gitreview for stable/train","message":"Update .gitreview for stable/train\n\nChange-Id: Icd3e100a4254a0f8b77e55290b8de912787c84a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/71ed2d2a2bcddacea7d3481ce2cefea3f2d5f6eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/71ed2d2a2bcddacea7d3481ce2cefea3f2d5f6eb"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
