)]}'
{"id":"openstack%2Fkuryr-libnetwork~683909","triplet_id":"openstack%2Fkuryr-libnetwork~master~I7f5ec495c91b79f65f73528600372405e2003899","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"I7f5ec495c91b79f65f73528600372405e2003899","subject":"Fix unit tests CI job","status":"MERGED","created":"2019-09-23 10:07:32.000000000","updated":"2019-10-02 00:34:29.000000000","submitted":"2019-10-01 09:13:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683909-1569921209974-02882d20","meta_rev_id":"615a6a9575db0efa5d77a1c25a95105402adc1f8","_number":683909,"virtual_id_number":683909,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 09:13:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2019-09-25 12:21:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2019-10-01 08:27:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2019-10-01 08:27:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"date":"2019-09-24 02:41:48.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"date":"2019-09-24 02:41:48.000000000","_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"date":"2019-09-24 02:41:48.000000000","_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-24 02:41:48.000000000","_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 02:41:48.000000000","updated_by":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"reviewer":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"state":"REVIEWER"},{"updated":"2019-09-24 02:41:48.000000000","updated_by":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"reviewer":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"state":"REVIEWER"},{"updated":"2019-09-24 02:41:48.000000000","updated_by":{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},"reviewer":{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},"state":"REVIEWER"},{"updated":"2019-09-24 02:41:48.000000000","updated_by":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"reviewer":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"state":"REVIEWER"},{"updated":"2019-09-25 12:21:24.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-10-01 08:27:42.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2019-10-01 09:13:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72d3b496f504ee91ce42bbb03df962069a59a179","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-09-23 10:07:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"206882738d157b35737c02b2301ac0980088c382","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 10:58:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f460535e3852464799ba1a5fd9aa9a0a : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e9d652dfd504d64820b015dcc7ad834 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b6161301df439a92df3b4ec05a06bb : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3e46f96f2ef454a99699a0160da0ee7 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b067facc83246c4972bf343769465eb : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a6d268312a34afdb62b5e97a1809920 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a30eeecef04348a35dd3688ccf8f88 : SUCCESS in 4m 26s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/2c53ec33e1704dd38098d594b62edcf7 : SUCCESS in 30m 10s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/843dccea70254c7684f902d3a9d008d0 : SUCCESS in 33m 52s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/bf8b30f52a5449d4ad7c78dbf2904ace : SUCCESS in 35m 58s\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/755dbbccc01e4cd5b6b19f6b68329431 : SUCCESS in 45m 01s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/936efa65e1ad4921ac7708cc870fbb33 : SUCCESS in 30m 33s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/15588d15287e432884f52fcfe4a7f4fd : SUCCESS in 46m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5270c122d42c540b86438f3ee7bc1ae040587df","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-09-25 12:21:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f07ddfbf21ed1fca36cd3ae9e86ae6b135df3d88","author":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"date":"2019-10-01 08:27:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a2d3e7eb2757653acf7fc972fdbb9e032b4b92f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 08:27:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db148ad5bd34211d1383c3cecf348bb45666aa81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 09:13:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e3ea60969b4d4ddc65e9176ab0d3fb98511750e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 09:13:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b29e976533c4999afc8ddc0ac6bd4a3 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73dbe7cc5f60400cabffeda28dd3eeb9 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/298736aee10b4fdd975fb78bb867eb82 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe9c9a7f86a44be594d0b0d4febbcd65 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd0b749a32df4bfebc6fd6c1a9488d4f : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c187a1f4b74254936c717d6b1233d5 : SUCCESS in 4m 09s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/b1ade6f476074603b199d447a55d43b3 : SUCCESS in 33m 51s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/57f3b205102445b88e969476e6019452 : SUCCESS in 33m 05s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/4c31a08ff1ae4c499fba03dfe22d409a : SUCCESS in 36m 30s","accounts_in_message":[],"_revision_number":1},{"id":"90f0f885db2becb69caa9e1ef8ec275b0f841b7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 09:15:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc84a8006c84aa88fe2d111824dd7aa : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1},{"id":"615a6a9575db0efa5d77a1c25a95105402adc1f8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-10-02 00:34:29.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 90847c601fbba38dcd7e237a98fbb6f91321a39a","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46442db3a249eb2276b83000e574a7b9fc256e58","revisions":{"46442db3a249eb2276b83000e574a7b9fc256e58":{"kind":"REWORK","_number":1,"created":"2019-09-23 10:07:32.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/09/683909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/09/683909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/09/683909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/09/683909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/09/683909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/09/683909/1"}}},"commit":{"parents":[{"commit":"622ded3e28c37616a667b2901a3fa2d2ce277de6","subject":"Merge \"Fix ipv6 issue on devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/622ded3e28c37616a667b2901a3fa2d2ce277de6"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-23 10:06:46.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-09-23 10:06:46.000000000","tz":0},"subject":"Fix unit tests CI job","message":"Fix unit tests CI job\n\nChange-Id: I7f5ec495c91b79f65f73528600372405e2003899\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/46442db3a249eb2276b83000e574a7b9fc256e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/46442db3a249eb2276b83000e574a7b9fc256e58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
