)]}'
{"id":"openstack%2Fkuryr-libnetwork~691588","triplet_id":"openstack%2Fkuryr-libnetwork~master~Ifeb2d1c713a7e80b7a5d7b8fc37095a42c180ea5","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"Ifeb2d1c713a7e80b7a5d7b8fc37095a42c180ea5","subject":"Stop testing python2.7","status":"ABANDONED","created":"2019-10-28 05:53:05.000000000","updated":"2019-11-19 14:44:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"03a2a00597786ff4219ae889e409718e8762c77f","_number":691588,"virtual_id_number":691588,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-28 06:44:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-1,"date":"2019-10-28 17:07:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 06:44:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 17:07:06.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"9a4e0c3ac2fda55a99584320ee7cf58949d36e6c","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-10-28 05:53:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9a1d47162df05fc58374d19e8ffb787764d0ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 06:44:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cc0f5722751a45a1a3c3b62b23d3ca43 : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8752775741f4e8a8070dd8665ebffc5 : SUCCESS in 3m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/212c4bf969554ff3832018f701bbd67e : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e4d5ae8fad94da09fca7292fc85849a : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dac073eb74264685b40f934253ef7eaa : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59bae9c953a04df68864a5bae591eddf : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/08ad544e9e7349499f2345c95cea3f6c : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f1c7f823f6f4aeab086365af3ec858c : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9c56376be9e4a3fabe962fab0d65623 : SUCCESS in 3m 09s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/7b890e73242f4139910a76ffe61e3130 : SUCCESS in 29m 21s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/54ea41d336314ccb82423cba6059609c : SUCCESS in 36m 57s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/5b7f8839602a426da3fd90c5d9d624d1 : SUCCESS in 30m 00s\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/960c952a734741c9b58d85003518ca2f : SUCCESS in 43m 47s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/a8ac085e47f04d76b59e063b9799910d : FAILURE in 38m 01s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/17b9f57d378a4bf5aa278038601cadfa : SUCCESS in 45m 15s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/42f4844a180549098e68636645c05f0c : SUCCESS in 44m 37s","accounts_in_message":[],"_revision_number":1},{"id":"61613c82e2819dd62a3413d842663b59bdb2dbc1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-28 17:07:06.000000000","message":"Patch Set 1: Code-Review-1\n\nwe need to do a lot more on this and as per discussed schedule. I am composing the community-wide goal  -https://review.opendev.org/#/c/691178/\n\nLet\u0027s be on sync on what all to do and when via goal.","accounts_in_message":[],"_revision_number":1},{"id":"03a2a00597786ff4219ae889e409718e8762c77f","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-11-19 14:44:43.000000000","message":"Abandoned\n\nAbandoning in favor of https://review.opendev.org/#/c/694449/.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e87ebf998d75cc5b6fdfb8290b930b58df1db92","revisions":{"6e87ebf998d75cc5b6fdfb8290b930b58df1db92":{"kind":"REWORK","_number":1,"created":"2019-10-28 05:53:05.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/88/691588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/88/691588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/88/691588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/88/691588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/88/691588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/88/691588/1"}}},"commit":{"parents":[{"commit":"fc7cc09604041e628d9a48777aad77f666dea8a1","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/fc7cc09604041e628d9a48777aad77f666dea8a1"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-10-28 05:52:43.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-10-28 05:52:43.000000000","tz":480},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nChange-Id: Ifeb2d1c713a7e80b7a5d7b8fc37095a42c180ea5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/6e87ebf998d75cc5b6fdfb8290b930b58df1db92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/6e87ebf998d75cc5b6fdfb8290b930b58df1db92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
