)]}'
{"id":"openstack%2Fkuryr-libnetwork~708053","triplet_id":"openstack%2Fkuryr-libnetwork~master~Id9158df88f0b6ed0c071994bf0aa240063ac6761","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"Id9158df88f0b6ed0c071994bf0aa240063ac6761","subject":"Update docs and scripts for using python3","status":"MERGED","created":"2020-02-17 04:46:05.000000000","updated":"2020-02-26 12:08:23.000000000","submitted":"2020-02-26 12:06:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708053-1582718803742-1d1e633c","meta_rev_id":"d50f1aa5127cb7a731e7707a4985ccb7ed0ee0d1","_number":708053,"virtual_id_number":708053,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"date":"2020-02-24 20:47:13.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 12:06:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-02-24 13:44:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":2,"date":"2020-02-26 11:27:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"date":"2020-02-26 00:49:04.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":1,"date":"2020-02-26 11:27:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"date":"2020-02-26 00:49:04.000000000","_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-26 00:49:04.000000000","_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"date":"2020-02-26 00:49:04.000000000","_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-24 13:44:10.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-02-26 00:49:04.000000000","updated_by":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"reviewer":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"state":"REVIEWER"},{"updated":"2020-02-26 00:49:04.000000000","updated_by":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"reviewer":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"state":"REVIEWER"},{"updated":"2020-02-26 00:49:04.000000000","updated_by":{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},"reviewer":{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},"state":"REVIEWER"},{"updated":"2020-02-26 00:49:04.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2020-02-26 11:27:51.000000000","updated_by":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"reviewer":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"state":"REVIEWER"},{"updated":"2020-02-26 12:06:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57edbf130974abd77cdbc669570411ba976d2a59","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 04:46:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4632f0cb44f586fca756f3941edcb45433b95110","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 04:55:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bbe91f49063ed77d56aaf46a03d13b34fb7755a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 05:10:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/82c07d4ef87645358f67ba4c291512cc : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03f602503aa14220a2a7255e708ff1e1 : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1d3add5afad4f4f947b96b8349b25da : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f29619d56a2414385427d1aafffa419 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1261b9f62e884f21b23d824ed5f4ecea : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e03c1081b6a14ba0bb74612a013d83b6 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c553f10de0c4787bde061582cc088c8 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b2499456af04c818b03f4a72b4fa6f2 : SUCCESS in 4m 52s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/64516fec73514fceb328b8d1ed6b9d83 : FAILURE in 10m 06s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/fe02dc3dfc074864adb78369818f0e8f : FAILURE in 14m 03s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/3656f67c0f6949599162f1015201ffc9 : FAILURE in 10m 40s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/9eed22ab9b5f445aa5bfe0d67a30390f : FAILURE in 11m 50s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/221acda2312447119f6408d464aee449 : FAILURE in 8m 23s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/f7e6b116aa85479c95b82b2ba7877f86 : FAILURE in 11m 43s","accounts_in_message":[],"_revision_number":2},{"id":"6d99672e643eeb281a754e65bef28f2baf064882","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 16:15:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d0e7f290d094e3ca1dfb57991fd43b9bb8d3d0ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 17:24:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/53071414de0648f5a61ad5d09ab4bfa8 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16f8cc1e7fa840749cf58836271d0dac : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc585d664c344db2a1f701fa589ba0eb : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2ab8d9925a043f985b5953ba70dfee9 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cfbd0a26aca43d48f7e1c6fe513c2ce : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35f2d0f7d3084d27bf21b1d24c1b2550 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/286abe33aa58436d8c8946bd7f90ae08 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc73590723e434aadd17c9da7fd6aca : SUCCESS in 6m 06s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/bb006e93ff4b4263978d9cc1540f158d : SUCCESS in 44m 54s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/e2437da215e145f2adba267cd31f51ad : FAILURE in 32m 48s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/e558f4f6f4af4089b9711b60b712c237 : SUCCESS in 56m 54s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/137e74a2f7ab4d8c8a7616c8adf425a7 : FAILURE in 43m 55s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/e427dda502f04382a60402d1e9a1b074 : SUCCESS in 52m 46s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/f4d0ddab9d864bf2af556c703113143b : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":3},{"id":"5d6b15f4b1f2307e4aed7b37b9f7b1de8161e040","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-18 04:36:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8417655e8ed1b2aeb8de79d2b9bb54d4d5052f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 05:30:16.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/307dcb6c43704b7b839828ee93d94a4d : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66ba76d134cd4171a5b00854a234ecd9 : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14bcccb253db4766b5dc5c9d4bea0e5f : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19cf2a989325427fbdd54441b68c5375 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7086769e83b74a2c88f553793989271b : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d570256005bc4bab9a246f282e82c1bc : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b511fc1bbc5f47f5a0a11ba8ee5b4c38 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4519cb58c6024e62b0ea37007a610ab2 : SUCCESS in 5m 04s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/80a23b953c9b4b8ab14ed89b28d1c9a1 : SUCCESS in 33m 11s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/45b6639ddc5c489d959fd0f582de2356 : FAILURE in 39m 41s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/4e540ce84ed7437fb2ff27ca24bf73a0 : SUCCESS in 51m 54s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/6fb65512afad4c54a6893850ae8d6fa8 : FAILURE in 38m 48s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/3c32bd9533144f21834e207d46bd50f1 : SUCCESS in 52m 04s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/47782be68c5c4911937932f5672c1477 : SUCCESS in 33m 42s","accounts_in_message":[],"_revision_number":3},{"id":"14fbd375ffbc67b26a9838b72953db5a34b448e4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-22 16:36:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1533054dd8e3c9f8214a47e2433a5b05fa1d5c42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 17:32:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e2551b22cc564951b8a8a2897a8a7358 : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cde6009eef55435faad47cf7b6865a31 : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de5dde39dfd54b9fb4c5de55f3555977 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5810ca24cb94f61b3938ab3d300ce37 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de7f5f1821784af6b1a9fe012f44e845 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ceaf031ef6424582af5237fdcec65426 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0645a56727c046a582b00e86a8a70f19 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091741e71f8641bbabb649b7e1019477 : SUCCESS in 4m 45s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/2a1c1b0ff6f443288de859b1474b6ddb : SUCCESS in 42m 17s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/1b60c19975934b4abb3e720e784e3488 : RETRY_LIMIT in 24s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/dc1f82d93d454cfc91dc0b3fea147c0b : RETRY_LIMIT in 12s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/c1e8a4719494431ea2096027b93f3239 : FAILURE in 44m 22s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/cc86d1268a4f46bc8bf4d59a5af0b1a8 : SUCCESS in 48m 44s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/b3a99bda324c449b9edb17b0af60a3bb : RETRY_LIMIT in 26s","accounts_in_message":[],"_revision_number":3},{"id":"25930ec18156fdbd3c5bba339ca595cc0d67c928","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-22 19:44:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"788083491b342c24d7fc506a98b4a47edc78d10f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 20:29:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3f7bf21d4a5f408a9e36abb4c1e8f11b : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6585e90fd62f45f089e5772434dfee80 : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc7e2918666a44b6a040658146d00458 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96853c26d3204a35891846a8771882cd : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f93fb86a21ce497f9b2fd4c7ee9fb9a5 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86dddff3105443ad85e19061b2717189 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/821adfbc199d4612afda4e768a00a33a : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fca15d68a7e4b0792d88a288d40f356 : SUCCESS in 4m 43s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/933a2508274e469c8a8d4feb3983773f : SUCCESS in 32m 18s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/e766c669d13243f082557005c8879c6e : FAILURE in 33m 38s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/d78f532034f44ebe97635a6c3d308d20 : SUCCESS in 41m 07s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/09e333ecffa34f798e7023e58dc6377b : FAILURE in 32m 35s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/7707a1dd3b0b4bfeaa32b16f0e516493 : SUCCESS in 42m 43s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/7831e66814df4d89b4366f51b16fe270 : SUCCESS in 33m 24s","accounts_in_message":[],"_revision_number":3},{"id":"c32212055dceb12bc25ce51826b9a6022fd1b522","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-24 13:44:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c71a8a0b3158d557e67e4caa7c42dbf8c2d52a5e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-24 20:47:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7e11e16e38d114299ffbe9512bab27f2b12098ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 21:35:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/72d8ca124e05431e9db43a16230bde7b : SUCCESS in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36fd8a1acde5463aa8f7eb1e1c97ad48 : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f7813357e3144a3b34946c41a5c811f : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe88dea7579c45ca850db31fe62e1c03 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7e16ef7b9e14d9290f5d121a2ae7c41 : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80857afd865b40da83ce73869e6593d8 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60ce4b13c3cc4d23807c3bc94120c477 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ebde9a9e77a438184e601a0d44cbb9d : SUCCESS in 5m 16s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/e64deb392d234956b7819e83a7cdfdfe : SUCCESS in 30m 26s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/d707cdb28af04ad69ee714a74e4d1155 : SUCCESS in 32m 52s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/00b09188d4a1461aa110a1b04a29482d : SUCCESS in 45m 18s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/6eb5f0d105be4998b8f0084db1deb47a : SUCCESS in 35m 06s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/260c849a62524ad58775f256196cc6c1 : SUCCESS in 44m 59s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/0884cceee18a44279148911d6b5db561 : SUCCESS in 34m 05s","accounts_in_message":[],"_revision_number":3},{"id":"ae9d14da059e0be95de92e0a429c82de482707ac","author":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"date":"2020-02-26 11:27:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"311723d95563a55e5ce01a3afe90547e97de1c97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 11:29:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"497f889be7aa700e1e3aadb21f477180cb39df62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:06:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/68df5249161b45fc89c5c11d1361c04a : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2ebc5b6d5874617a7552cbd739cdcc0 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f23c76b55ec4fd5875209fdbcc44a49 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a4d288d188b40528e93f99590b707c8 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0675d8d2bf24fff9342f38a63fb860a : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eaeb16f0c824b5b896e8ec202be3d8c : SUCCESS in 5m 13s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/e81f955042fd47869f59bcd8bd0fd7b9 : SUCCESS in 35m 24s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/69b46b05f8f84a71a586347d9fe8ce28 : SUCCESS in 36m 07s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/d68a75ca02534d9ca29b68b45f440ef3 : SUCCESS in 33m 29s","accounts_in_message":[],"_revision_number":3},{"id":"0eaa8519a29fed795022d8e7ef3910e87fec6ce4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:06:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d50f1aa5127cb7a731e7707a4985ccb7ed0ee0d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:08:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c9b23bc41b544048e72b3f257303405 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d9523c7afe55118fa463b515fe60f99104551210","revisions":{"ab691b14ea387292c751fb196c781a440eece254":{"kind":"REWORK","_number":1,"created":"2020-02-17 04:46:05.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/53/708053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/53/708053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/708053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/708053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/708053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/708053/1"}}},"commit":{"parents":[{"commit":"97266ff5b904db5bd89930ff8fa7bc4ba3506ebe","subject":"Retry a few times on listing extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/97266ff5b904db5bd89930ff8fa7bc4ba3506ebe"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 04:45:44.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 04:46:04.000000000","tz":0},"subject":"Update installation guide for using python3","message":"Update installation guide for using python3\n\nChange-Id: Id9158df88f0b6ed0c071994bf0aa240063ac6761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/ab691b14ea387292c751fb196c781a440eece254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/ab691b14ea387292c751fb196c781a440eece254"}]},"branch":"refs/heads/master"},"ae13d730537122160cfcd2e0c4bd579d142ec0b8":{"kind":"REWORK","_number":2,"created":"2020-02-17 04:55:09.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/53/708053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/53/708053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/708053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/708053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/708053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/708053/2"}}},"commit":{"parents":[{"commit":"97266ff5b904db5bd89930ff8fa7bc4ba3506ebe","subject":"Retry a few times on listing extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/97266ff5b904db5bd89930ff8fa7bc4ba3506ebe"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 04:45:44.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 04:53:43.000000000","tz":0},"subject":"Update docs and scripts for using python3","message":"Update docs and scripts for using python3\n\nPython2 is deprecated so we need to switch to python3\n\nChange-Id: Id9158df88f0b6ed0c071994bf0aa240063ac6761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/ae13d730537122160cfcd2e0c4bd579d142ec0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/ae13d730537122160cfcd2e0c4bd579d142ec0b8"}]},"branch":"refs/heads/master"},"d9523c7afe55118fa463b515fe60f99104551210":{"kind":"REWORK","_number":3,"created":"2020-02-17 16:15:17.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/53/708053/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/53/708053/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/708053/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/708053/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/708053/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/708053/3"}}},"commit":{"parents":[{"commit":"97266ff5b904db5bd89930ff8fa7bc4ba3506ebe","subject":"Retry a few times on listing extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/97266ff5b904db5bd89930ff8fa7bc4ba3506ebe"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 04:45:44.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 16:15:11.000000000","tz":0},"subject":"Update docs and scripts for using python3","message":"Update docs and scripts for using python3\n\nPython2 is deprecated so we need to switch to python3\n\nChange-Id: Id9158df88f0b6ed0c071994bf0aa240063ac6761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/d9523c7afe55118fa463b515fe60f99104551210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/d9523c7afe55118fa463b515fe60f99104551210"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
