)]}'
{"id":"openstack%2Fkuryr-libnetwork~709319","triplet_id":"openstack%2Fkuryr-libnetwork~master~Iecd61c0ba0c9ae789b3cf771dc435d7b9513a339","project":"openstack/kuryr-libnetwork","branch":"master","topic":"bug/1864330","hashtags":[],"change_id":"Iecd61c0ba0c9ae789b3cf771dc435d7b9513a339","subject":"Fix test_container_ipam_request_address_with_existing_port","status":"MERGED","created":"2020-02-22 20:07:26.000000000","updated":"2020-02-24 14:34:52.000000000","submitted":"2020-02-24 14:32:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709319-1582554761598-795c03e7","meta_rev_id":"d199e231d76e93936304c8e888be3845e6920d91","_number":709319,"virtual_id_number":709319,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-24 14:32:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":2,"date":"2020-02-24 13:39:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":2,"date":"2020-02-23 12:17:35.000000000","_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":0,"date":"2020-02-22 22:33:05.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":1,"date":"2020-02-24 13:39:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"date":"2020-02-22 22:33:05.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-22 22:33:05.000000000","_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"value":0,"date":"2020-02-22 22:33:05.000000000","_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-22 22:33:05.000000000","updated_by":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"reviewer":{"_account_id":9820,"name":"Liping Mao","email":"limao@cisco.com","username":"LipingMao"},"state":"REVIEWER"},{"updated":"2020-02-22 22:33:05.000000000","updated_by":{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},"reviewer":{"_account_id":11343,"name":"Gal Sagie","email":"gal.sagie@gmail.com","username":"GalSagie"},"state":"REVIEWER"},{"updated":"2020-02-22 22:33:05.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2020-02-22 22:33:05.000000000","updated_by":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"reviewer":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"state":"REVIEWER"},{"updated":"2020-02-23 12:17:35.000000000","updated_by":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"reviewer":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"state":"REVIEWER"},{"updated":"2020-02-24 13:39:49.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-02-24 14:32:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5746a7cce03a4e93da01333a2fe8d5e5b873439","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-22 20:07:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1681b63ed2ccf8e7e4aff925cdb2ee5d7c623af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 20:53:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06a60913046f454bb208d70c81ca99a2 : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/300f5597525947c7a3f0b4f805951833 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/862cc80ae6c24790886a34679a6ad219 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4123fc0a2d94140a9069b9e0170c421 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b94908371a64cd0bfc9d9cc0b8e9492 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f1a2a131a07410fb30efb5fe241b5be : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac7227a6ab344348e3273ad7652b4b1 : SUCCESS in 5m 04s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/a2dee0903daa43ca851042ff0c198bac : SUCCESS in 30m 11s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/dabddc3a740040538fb295841336d794 : SUCCESS in 32m 35s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/79eb514215274bf2888b34ce6e5312ae : SUCCESS in 44m 24s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/e8dc67c6af834f1ca7169c22cfe9d427 : SUCCESS in 33m 48s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/65764ca95b244f32835bb079354fb511 : SUCCESS in 45m 15s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/bf55d2739f68484fb2a0599f905c3aa1 : SUCCESS in 32m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f4b7bc1b1e4a2847ce1bb7bec40dc37ce89114d4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-22 21:33:34.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b2edfc4e72ddf37b7c0e07a207708181731fb50c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 22:29:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/333e4d67c2724135b860311e38fba5dd : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/962c79a8b30844bcb4c91592b547eaf8 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0512820fb47e4df5994e56e2b6687fb4 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/953e84fc132840fcb7263e91adfaf397 : SUCCESS in 9m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/020f7d51cd704715ae46ae9e0f3e4188 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14f85be5e55a4f2ca2e9c3801541ff11 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0148641ec5485ba6e92e75ed7351b3 : SUCCESS in 4m 46s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/f6ff4565084741b685ab472a3c2acc3a : SUCCESS in 28m 16s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/e5b4d7ddd4744780819f3175e6aae194 : SUCCESS in 37m 17s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/9286191d60374e558b48dbb0645b653b : SUCCESS in 44m 21s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/19f05b3b42254d26aadfd43c337f3b0b : SUCCESS in 32m 10s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/5637d0a05e4f46f2a7e6e835daedb2bc : SUCCESS in 49m 13s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/bafc16a1d9a74dc88775f0bf229c3c93 : SUCCESS in 37m 50s","accounts_in_message":[],"_revision_number":2},{"id":"8fa33a918d37ab7dd2d4f9b7fc132f6717a6b359","author":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"date":"2020-02-23 12:17:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9be11f957c52641e2c99f187693ec5a186879ebc","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-24 13:39:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c3f26722a522ff8f4dcc37fd733206a21a9acf0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 13:40:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4bc537a3b4d7ea332e2b3dca5c30f5f3e0b8356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 14:32:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3683e84839a24f109c1d6426bbef6a6c : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/271093a6b3a244539ac903610428d697 : SUCCESS in 9m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65e4bec4377c44d2b34c8aaa5662478d : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8eb7a8ecd784166a8343b41be027163 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/524e46254ed6449094c30fea2e358ce5 : SUCCESS in 5m 14s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/347d9d2208a640408cce9138197c5d44 : SUCCESS in 44m 10s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/3d4c17b4cb044f39a8c5df53114d7927 : SUCCESS in 39m 12s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/fe0fba16a3a04d4b836eb7413448bb2a : SUCCESS in 38m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4b2f4f272d23f31d8e92d34c978523bd59997b5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 14:32:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8aede3f9b66b6d02650d4961b5a8d1f23fe4f85a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 14:34:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6051f4a7abf042bc86d0abc68e1268da : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"62ff2f9f58d3dde3f66b0e96deed3f5431b32ce3","revisions":{"8f6a7b9f7b0dc52b91309ca3545874905d23a7fc":{"kind":"REWORK","_number":1,"created":"2020-02-22 20:07:26.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/19/709319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/19/709319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/19/709319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/19/709319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/19/709319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/19/709319/1"}}},"commit":{"parents":[{"commit":"97266ff5b904db5bd89930ff8fa7bc4ba3506ebe","subject":"Retry a few times on listing extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/97266ff5b904db5bd89930ff8fa7bc4ba3506ebe"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-22 20:06:57.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-22 20:06:57.000000000","tz":0},"subject":"[DO NOT MERGE] Test CI - 2","message":"[DO NOT MERGE] Test CI - 2\n\nChange-Id: Iecd61c0ba0c9ae789b3cf771dc435d7b9513a339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8f6a7b9f7b0dc52b91309ca3545874905d23a7fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8f6a7b9f7b0dc52b91309ca3545874905d23a7fc"}]},"branch":"refs/heads/master"},"62ff2f9f58d3dde3f66b0e96deed3f5431b32ce3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-22 21:33:34.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/19/709319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/19/709319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/19/709319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/19/709319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/19/709319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/19/709319/2"}}},"commit":{"parents":[{"commit":"97266ff5b904db5bd89930ff8fa7bc4ba3506ebe","subject":"Retry a few times on listing extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/97266ff5b904db5bd89930ff8fa7bc4ba3506ebe"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-22 20:06:57.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-22 21:33:00.000000000","tz":0},"subject":"Fix test_container_ipam_request_address_with_existing_port","message":"Fix test_container_ipam_request_address_with_existing_port\n\nCloses-Bug: #1864330\nChange-Id: Iecd61c0ba0c9ae789b3cf771dc435d7b9513a339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/62ff2f9f58d3dde3f66b0e96deed3f5431b32ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/62ff2f9f58d3dde3f66b0e96deed3f5431b32ce3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
