)]}'
{"id":"openstack%2Fkuryr-libnetwork~717742","triplet_id":"openstack%2Fkuryr-libnetwork~master~Ib7c6b7437abc0b5c500a5afdaaf79a045f7df9fe","project":"openstack/kuryr-libnetwork","branch":"master","topic":"fix-docs","hashtags":[],"change_id":"Ib7c6b7437abc0b5c500a5afdaaf79a045f7df9fe","subject":"Fix doc building","status":"MERGED","created":"2020-04-06 12:48:50.000000000","updated":"2020-04-06 18:55:32.000000000","submitted":"2020-04-06 18:53:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717742-1586199220749-9e742fff","meta_rev_id":"4eb88639da2eebc469b991af1c4922d2b06e7621","_number":717742,"virtual_id_number":717742,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-06 18:53:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-04-06 15:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2020-04-06 18:06:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-04-06 18:06:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"date":"2020-04-06 15:09:56.000000000","_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-06 15:09:46.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-04-06 15:09:56.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2020-04-06 18:06:47.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-04-06 18:53:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22452db80a6c6bdfb92fcb00961e9666991f4cc5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-06 12:48:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b92e6d6dc48dbcf593196109e6d4d91cd95bfe79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 13:33:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e76c1bdf6324fc793a8c26f683afb35 : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7b206f2334141338c90c49627a87574 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9b51333f7f48acb475203c444ea3a5 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ac8aa79fe44fb497bc8b7a8a0dd309 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17e17cf291134bdf85c4824e3dacdd74 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6309225fec364abfb08d0c5f638d82f0 : SUCCESS in 6m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0f6fa2741c8419b8356a344fd060c6b : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472876353ec54fad80ee9c1d18ab1dcf : SUCCESS in 4m 19s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/b9ed2543b25348879a623f7b55ec7808 : SUCCESS in 30m 41s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/e05e8d21ef5e4868bb19ce6fcfc83475 : SUCCESS in 37m 05s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/3f3bb38e6f4c4a86884ff9f923206f68 : FAILURE in 35m 07s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/adf0e8a856aa48cdb0469e1c56723de4 : SUCCESS in 32m 17s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/83a8535b78cb41c2b0f06946c4e46359 : FAILURE in 35m 44s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/58d6118ac2194e7bb9520113ab2c778a : SUCCESS in 32m 26s\n\nWarning:\n  Comments left for invalid file kuryr_libnetwork/tests/unit/port_driver/test_driver.py\n  Comments left for invalid file kuryr_libnetwork/port_driver/drivers/sriov.py","accounts_in_message":[],"_revision_number":1},{"id":"780a097541ade934eec861bccd952a6c774d39d4","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-04-06 15:09:46.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks okay!","accounts_in_message":[],"_revision_number":1},{"id":"0f9d657e2f1bb498b6964eb63f96df7fb72757b6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-04-06 18:06:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5687262ba824c5523c5409842e5d86c857fbef69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 18:06:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37af58e81ae9e2c19a462e8efea225ed6e7fb748","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 18:53:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58f3e3b54dc947d78970e162bbd2ae52 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dad233daa5b64ec08317998a8caac4e8 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/369fc3f9105840b3a89d94ebd568a7e8 : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc8d46e60d7a4a8e9936dfc14de6a5d7 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbccfabbef4542b8b204ccb2f5a91779 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac32867d16944a66b9aa857e092237ff : SUCCESS in 4m 15s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/213423b3b3734939a4827794f13520c3 : FAILURE in 39m 49s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/96d33030a8264d75aaaa1abd0b64a4bf : SUCCESS in 32m 55s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/36d9b0c65cdb44b2acf2c2fcb6206dc5 : SUCCESS in 31m 57s\n\nWarning:\n  Comments left for invalid file kuryr_libnetwork/tests/unit/port_driver/test_driver.py\n  Comments left for invalid file kuryr_libnetwork/port_driver/drivers/sriov.py","accounts_in_message":[],"_revision_number":1},{"id":"065a411f1a55dd701ac6ca5c72c390cb1a671e4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 18:53:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5f8e072fe62c38cd9b9a7c05270c3b79a46f6224","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 18:55:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d60c36c9c5044af9114151da0e44a97 : SUCCESS in 1m 29s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae6ae1a717945e1b9dda40b4be33dee : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d63d664b284923b27c4091563352e2dae0531be2","revisions":{"d63d664b284923b27c4091563352e2dae0531be2":{"kind":"REWORK","_number":1,"created":"2020-04-06 12:48:50.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/42/717742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/42/717742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/42/717742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/42/717742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/42/717742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/42/717742/1"}}},"commit":{"parents":[{"commit":"bfbaea9749ee157ec0dea782abb6b7f0a5c4ff98","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/bfbaea9749ee157ec0dea782abb6b7f0a5c4ff98"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-06 12:47:35.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-06 12:48:39.000000000","tz":120},"subject":"Fix doc building","message":"Fix doc building\n\nUse sphinx-build, this fixes doc building so that the generated content\nis available for publishing again.\n\n(change I0dbb13d555eb473916ba585c62a0f7433d15afed broke this)\n\nChange-Id: Ib7c6b7437abc0b5c500a5afdaaf79a045f7df9fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/d63d664b284923b27c4091563352e2dae0531be2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/d63d664b284923b27c4091563352e2dae0531be2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
