)]}'
{"id":"openstack%2Fkuryr-libnetwork~722073","triplet_id":"openstack%2Fkuryr-libnetwork~stable%2Fussuri~I79df54cecdc43ac889f1954ce1c6d19922e0e03c","project":"openstack/kuryr-libnetwork","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I79df54cecdc43ac889f1954ce1c6d19922e0e03c","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-22 16:25:06.000000000","updated":"2020-04-23 08:44:15.000000000","submitted":"2020-04-23 08:40:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722073-1587631241855-6b5ca73b","meta_rev_id":"83e40fec67d3722e710afd0c8e33d1f3ececebdc","_number":722073,"virtual_id_number":722073,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 08:40:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-04-22 18:50:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2020-04-23 07:09:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-04-23 07:09:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 18:50:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-04-23 07:09:31.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-04-23 08:40:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93e9845ed5021a74d0886622be796c1b0e1f1421","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-22 16:25:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f26a34af4208bc873398a58ff430acef52e6694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 17:25:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05572ed08af0415c9679469cbd8da4af : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ed3df4534b84a49a0b880e49b4c3852 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bb62aaba35f459dafcba35951bbdae0 : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1315e0ce61ba4498931f3c52d606cbdc : SUCCESS in 7m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22eb506b974849929156f5ab1b9f40ca : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e1a6037d3474e3d94468a1940952213 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1d34eb219c443df854859131a0c1951 : SUCCESS in 7m 43s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/a4b6d4ca6cab4d629dd9a46e0d40cde1 : SUCCESS in 42m 12s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/4916769963b046088a9ff565b2c5d22b : SUCCESS in 45m 06s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/c8cc8b7a4fb84c8c9afa016a305e9e1e : SUCCESS in 48m 50s","accounts_in_message":[],"_revision_number":1},{"id":"67573ef4641366f947d7b6b6f7b69a25d7095ee4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-04-22 18:50:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"94c887d351b9c0c993ee1627e2b93be8378b7a12","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-04-23 07:09:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6cb0ab0a0dd31ce0c2dccb70e4377655b758d8cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 07:09:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57f87ad7b0db2f40bd4a66111b2ad7c778a565bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 08:40:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8a1f0e05b3d3929e07de86b2d934adf5879a3b7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 08:40:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe8d4d6d3397409f84faa7d3e839beba : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889d3d24d6e04ab7929ac37b75fe8a4d : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/027ebc0337d24c109fc0d051adc70028 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3916b5918d5f42ba99956ee55c22d9ae : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d1d641f3a48473cb80a5c387a1aa711 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f36a8bc4bc84338a531547e838a701c : SUCCESS in 6m 48s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/b0ac419120db4db293e14b26918b2738 : SUCCESS in 33m 26s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/a592381ddf444a9db8fb3eb80eaf351c : SUCCESS in 37m 36s","accounts_in_message":[],"_revision_number":1},{"id":"cc285b872d1df6074ca348bd694ea752ced0d236","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 08:44:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a19a96f5f8649c0a8ca46cd5e9a5017 : SUCCESS in 2m 37s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2435ee4a937b456d92154983896f4cdb : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"080cd178566f2aba82aa2548c6e2ddacda4be054","revisions":{"080cd178566f2aba82aa2548c6e2ddacda4be054":{"kind":"REWORK","_number":1,"created":"2020-04-22 16:25:06.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/73/722073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/73/722073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/73/722073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/73/722073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/73/722073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/73/722073/1"}}},"commit":{"parents":[{"commit":"86a2816bee490856d7f88fe332c274a3810ad9fd","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/86a2816bee490856d7f88fe332c274a3810ad9fd"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 16:25:04.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 16:25:04.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I79df54cecdc43ac889f1954ce1c6d19922e0e03c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/080cd178566f2aba82aa2548c6e2ddacda4be054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/080cd178566f2aba82aa2548c6e2ddacda4be054"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
