)]}'
{"id":"openstack%2Fkuryr-libnetwork~728157","triplet_id":"openstack%2Fkuryr-libnetwork~master~I71b869fc509cb46b55b1e88fae20422bcba085f7","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"I71b869fc509cb46b55b1e88fae20422bcba085f7","subject":"Update hacking for Python3","status":"MERGED","created":"2020-05-14 17:24:23.000000000","updated":"2020-05-25 16:56:35.000000000","submitted":"2020-05-25 16:55:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728157-1590425715283-cdbf8240","meta_rev_id":"fdbabdaca46f5d1436d1b26c92db37260d962ed9","_number":728157,"virtual_id_number":728157,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-25 16:55:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-05-23 00:41:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2020-05-25 16:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-05-25 16:12:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 17:06:56.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-05-23 00:41:29.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-05-25 16:12:01.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-05-25 16:55:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b99f46bbf705aacc00fdff0804301502c877ce6","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-14 17:24:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a980873d8f3eed3ee6959ecd2d88de6d6ed853c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 18:11:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89ece1b99e0a4b07a6f1377060495a72 : SUCCESS in 9m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30cff00a2e1345ffbb98edc9f189e016 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a69eb7819ee41f086d50fde9f7700d7 : FAILURE in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbf8dc12ad234476aa4e8e1ecf76d600 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f25b8dcef1c4a8092a793f27ba32297 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57afc5226f1d413c9acc62cd03d42612 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24e0f4fcaba04328ad244aebe9921e9d : SUCCESS in 8m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5af6be0982db4010a3fe3c1267cec578 : SUCCESS in 2m 51s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/afd89e6dffce4f1f91d8ab4d63a61d34 : SUCCESS in 31m 11s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/fb19f08a0896452b85c5aa7d73645bea : SUCCESS in 37m 49s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/57a09e61db39442a988cbcac32993601 : FAILURE in 37m 09s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/66a193f52b934ec1ba0a464644ca7a42 : SUCCESS in 34m 22s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/2f5d3fe5bea8412eb37e72417615e372 : FAILURE in 40m 35s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/83536ae5fea643468432df46800f2427 : SUCCESS in 33m 54s\n\nWarning:\n  Comments left for invalid file kuryr_libnetwork/tests/unit/port_driver/test_driver.py\n  Comments left for invalid file kuryr_libnetwork/controllers.py\n  Comments left for invalid file kuryr_libnetwork/port_driver/drivers/sriov.py\n  Comments left for invalid file kuryr_libnetwork/tests/unit/base.py\n  Comments left for invalid file kuryr_libnetwork/opts.py\n  Comments left for invalid file kuryr_libnetwork/utils.py\n  Comments left for invalid file kuryr_libnetwork/port_driver/driver.py\n  Comments left for invalid file kuryr_libnetwork/tests/fullstack/test_ipam.py","accounts_in_message":[],"_revision_number":1},{"id":"7cfa12c404df8f6ac1b29e8f17b8af76251d27ec","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-15 16:42:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e5e894a0c9a3d2eeb244063d95c9797aa50275d0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-15 17:06:56.000000000","message":"Patch Set 2: Code-Review-1\n\nlet\u0027s remove the hacking and flake8, its plugins dep from lower-constraints.txt","accounts_in_message":[],"_revision_number":2},{"id":"93d8051cbce2a444aba42c8fc3a31333a3cae9f4","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-15 17:26:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"228b324adfedb8c5cbafd478c2bdab055b8f1325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 18:07:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00c880131dc04714a27a295060bbe6eb : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f534418d5b94e96a3d579d80dd1ce0a : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63e17200094e45b2b974a64b156f72ed : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef8f07073bb543a281801ebc22a8dcfd : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99fabe3ecfbb4cb79c2f78243dfd8bbd : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96fe70ebb9694b9eaa7c5a273b991388 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d6930e70b3f4a26bbd3f7ba59a7be06 : SUCCESS in 5m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fd049d455f3b4411873090b118710ca3 : SUCCESS in 3m 05s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/ca2f53b634434121bc8b6efdc65de577 : SUCCESS in 31m 10s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/a91012628e294e5ea77e7afbc777bd82 : SUCCESS in 34m 20s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/cf6bb781c9cd41309d4a2d51e616832b : FAILURE in 32m 18s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/d1a6c3767fcc48108dc08debb79c2ec1 : SUCCESS in 36m 24s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/9bd01611acfb4ed79d2b0e11a8634358 : FAILURE in 32m 10s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/96b287a685354079852ad2d7b9738c09 : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":3},{"id":"dca9b69dccb460a354894173eb38a27dd89d4aad","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-19 00:13:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9ed501638ca44498ef4a14e88171287e7799ce83","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-20 15:41:14.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nPlease fix Hongbin\u0027s remarks.","accounts_in_message":[],"_revision_number":3},{"id":"a69cc61e31063314a812e92e98726c1662ce8041","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-23 00:03:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7f6fe122b3cb47c2c7efae2041405ee95ace124c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-23 00:05:49.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8e19442044743d46ed7667bce1af8c11d9794186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 00:37:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd642c06f7674ebdba51cbcd9080e64f : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66b8c80c724f40d79b6e38b9b52d1515 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/294adc885d724373ac4a0aa624f5ba4d : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdca80691d894c28b30f2f856e5f6fa1 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf6ec19c55814b62bcc2236eca0b0c74 : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8c6a35a96954f418a572a1b63720edf : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3830262b1be54c6aaad712f1de3ca5ea : SUCCESS in 4m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b5d5692d0cb548c19d6d9c475870d042 : SUCCESS in 2m 12s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/64f5612266e544db911a60d407b3cc6e : SUCCESS in 27m 14s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/e45ba201002f422eb7d1d4317f3151b6 : SUCCESS in 31m 42s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/896c315722ce4d7d98a686c0da8aad2f : FAILURE in 28m 49s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/5a61c5030ce54864b13cb0de2be7bdbe : SUCCESS in 31m 59s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/bdc9474fd94c4539a54143ad7b7aea4b : FAILURE in 29m 05s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/855a41efb6b04e62bff16d7e097b72a6 : SUCCESS in 29m 13s","accounts_in_message":[],"_revision_number":4},{"id":"b5bdfff242dce7c034a48359f50ccaf29cc7ab03","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-23 00:41:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4497cdaaf2568bf296431ec17713236ed00886ee","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-25 16:12:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f4c47ecdd450add3ee301fa066a2ab72bb1990e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 16:12:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2a7ef1f5885f116e32390ec4310815001b4e55b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 16:55:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8959141914d2ddbe7b5475c77230e980acd8aae6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 16:55:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e53fa68998c4691a7eef2fc337bacf8 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a125b99a4584de9b220d586bdd95c5f : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82f68e98396d4a9c947da8b9a7971d3b : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b7321b1505d4e08bcdfeda4920667b2 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a675468474e64dbe89fdde6bc3d61700 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6248965bd87a42fc8c0e5092a7cd6972 : SUCCESS in 4m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e030e6eb632a462f8c9d77428addf0b5 : SUCCESS in 2m 11s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/5df06b3635ff4b69a320331348ddacb8 : SUCCESS in 37m 53s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/0efccc38ff1943269e8cda91599b5f52 : SUCCESS in 30m 02s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/c243ac2905dc40f4b7540475e4134c6c : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":4},{"id":"fdbabdaca46f5d1436d1b26c92db37260d962ed9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 16:56:35.000000000","message":"Patch Set 4:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/883d24ddf91e45d8a3c0f6d2ebb56a27 : SUCCESS in 1m 05s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba6e05f085241d9afcee0b0c583a003 : FAILURE in 1m 05s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"036703c8d4ae66595cbbbfc17b1afe8c0c46a7b2","revisions":{"24c49c3d2ba764ac5b1c764b3a5471ff0f1dc65f":{"kind":"REWORK","_number":1,"created":"2020-05-14 17:24:23.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/57/728157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/57/728157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/1"}}},"commit":{"parents":[{"commit":"afecf766f58067485bedd356d5ad986baf3dbe6c","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/afecf766f58067485bedd356d5ad986baf3dbe6c"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-14 17:23:50.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-14 17:23:50.000000000","tz":480},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found.\n\nChange-Id: I71b869fc509cb46b55b1e88fae20422bcba085f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/24c49c3d2ba764ac5b1c764b3a5471ff0f1dc65f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/24c49c3d2ba764ac5b1c764b3a5471ff0f1dc65f"}]},"branch":"refs/heads/master"},"4f619f497307672c65c48d405665db4ac610e34a":{"kind":"REWORK","_number":2,"created":"2020-05-15 16:42:05.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/57/728157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/57/728157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/2"}}},"commit":{"parents":[{"commit":"afecf766f58067485bedd356d5ad986baf3dbe6c","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/afecf766f58067485bedd356d5ad986baf3dbe6c"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-14 17:23:50.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-15 16:41:32.000000000","tz":480},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found.\n\nChange-Id: I71b869fc509cb46b55b1e88fae20422bcba085f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/4f619f497307672c65c48d405665db4ac610e34a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/4f619f497307672c65c48d405665db4ac610e34a"}]},"branch":"refs/heads/master"},"87d420be731865230f5d5bd084659f3bf01546e7":{"kind":"REWORK","_number":3,"created":"2020-05-15 17:26:39.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/57/728157/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/57/728157/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/3"}}},"commit":{"parents":[{"commit":"afecf766f58067485bedd356d5ad986baf3dbe6c","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/afecf766f58067485bedd356d5ad986baf3dbe6c"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-14 17:23:50.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-15 17:25:58.000000000","tz":480},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found.\n\nChange-Id: I71b869fc509cb46b55b1e88fae20422bcba085f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/87d420be731865230f5d5bd084659f3bf01546e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/87d420be731865230f5d5bd084659f3bf01546e7"}]},"branch":"refs/heads/master"},"036703c8d4ae66595cbbbfc17b1afe8c0c46a7b2":{"kind":"REWORK","_number":4,"created":"2020-05-23 00:03:51.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/57/728157/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/57/728157/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/57/728157/4"}}},"commit":{"parents":[{"commit":"5c03de1668315a5402861aeffc41a8235fb73ceb","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/5c03de1668315a5402861aeffc41a8235fb73ceb"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-14 17:23:50.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-05-23 00:03:51.000000000","tz":0},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found.\n\nChange-Id: I71b869fc509cb46b55b1e88fae20422bcba085f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/036703c8d4ae66595cbbbfc17b1afe8c0c46a7b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/036703c8d4ae66595cbbbfc17b1afe8c0c46a7b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
