)]}'
{"id":"openstack%2Fkuryr-libnetwork~729131","triplet_id":"openstack%2Fkuryr-libnetwork~master~I37f2acd77a953c67e18f0772cd1aa3d47e62f6a0","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"I37f2acd77a953c67e18f0772cd1aa3d47e62f6a0","subject":"Add releasenotes link to README","status":"MERGED","created":"2020-05-19 06:29:11.000000000","updated":"2020-05-26 12:37:40.000000000","submitted":"2020-05-26 12:32:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729131-1590496350869-3f6755ef","meta_rev_id":"e714cb5d408b628fc0a6335f057b22f09925f1d3","_number":729131,"virtual_id_number":729131,"owner":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 12:32:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-05-26 12:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2020-05-26 12:17:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 12:17:16.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-05-26 12:32:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df47d5c743efc5b33dce47d25a9c115e5e8a290a","author":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"date":"2020-05-19 06:29:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc524a5f0d7e5b8d3303af2346e92aafc46d6038","author":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"date":"2020-05-19 06:30:11.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7d8c57dcfb5669fc33e99fd0bdd4f7861f3d9d55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 07:14:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/484d5884c3964f9eb0dfa81efb54fbf3 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/714d6c21f08c4bf4be05db170c60b1ff : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a8508a76fc491fa8d5122367ba0e98 : SUCCESS in 5m 54s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/67a96936164a4eab9da334c19b2168d8 : SUCCESS in 33m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3859ea99ed8e9985037802cb9ca34276e6042a53","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-05-26 12:17:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7644a5271dd338e732890cd186e0bb1bc3a5f982","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 12:17:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eaf8eb6a608ff305493867ce31a9d9ac9ebc83a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 12:32:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/86ec7ef66b0e4da5aa321a8a9ea73c8a : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b519ac4840545688a325f37df3566a7 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/775a4fadac0e4788939cb06cc0084e3b : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":2},{"id":"28e4ea22cf611280ea8f49e7c2bd720e78a58edf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 12:32:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e714cb5d408b628fc0a6335f057b22f09925f1d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 12:37:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad732b584f334b63b475187148205953 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fcdb395e76ab57eb0b921b8f097a2835870168bf","revisions":{"ad3de9d7b1d8e37c4fa01cc914a91d62ea87abd6":{"kind":"REWORK","_number":1,"created":"2020-05-19 06:29:11.000000000","uploader":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"ref":"refs/changes/31/729131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/31/729131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/31/729131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/31/729131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/31/729131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/31/729131/1"}}},"commit":{"parents":[{"commit":"085fb3c9c58ef9193bb9088f9a7a970c2bacde24","subject":"Remove translation sections from setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/085fb3c9c58ef9193bb9088f9a7a970c2bacde24"}]}],"author":{"name":"lidong","email":"lidongbj@inspur.com","date":"2020-05-19 06:27:35.000000000","tz":480},"committer":{"name":"lidong","email":"lidongbj@inspur.com","date":"2020-05-19 06:29:02.000000000","tz":480},"subject":"Add releasenotes to README","message":"Add releasenotes to README\n\nChange-Id: I37f2acd77a953c67e18f0772cd1aa3d47e62f6a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/ad3de9d7b1d8e37c4fa01cc914a91d62ea87abd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/ad3de9d7b1d8e37c4fa01cc914a91d62ea87abd6"}]},"branch":"refs/heads/master"},"fcdb395e76ab57eb0b921b8f097a2835870168bf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-19 06:30:11.000000000","uploader":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"ref":"refs/changes/31/729131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/31/729131/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/31/729131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/31/729131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/31/729131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/31/729131/2"}}},"commit":{"parents":[{"commit":"085fb3c9c58ef9193bb9088f9a7a970c2bacde24","subject":"Remove translation sections from setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/085fb3c9c58ef9193bb9088f9a7a970c2bacde24"}]}],"author":{"name":"lidong","email":"lidongbj@inspur.com","date":"2020-05-19 06:27:35.000000000","tz":480},"committer":{"name":"lidong","email":"lidongbj@inspur.com","date":"2020-05-19 06:30:05.000000000","tz":0},"subject":"Add releasenotes link to README","message":"Add releasenotes link to README\n\nChange-Id: I37f2acd77a953c67e18f0772cd1aa3d47e62f6a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/fcdb395e76ab57eb0b921b8f097a2835870168bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/fcdb395e76ab57eb0b921b8f097a2835870168bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
