)]}'
{"id":"openstack%2Fkuryr-libnetwork~730953","triplet_id":"openstack%2Fkuryr-libnetwork~master~Ib5308485e50cc076855cc8e1df72b02d208febe1","project":"openstack/kuryr-libnetwork","branch":"master","topic":"cap_jsonschema","hashtags":[],"change_id":"Ib5308485e50cc076855cc8e1df72b02d208febe1","subject":"Cap jsonschema 3.2.0 as the minimal version","status":"MERGED","created":"2020-05-26 19:52:13.000000000","updated":"2020-06-03 16:39:01.000000000","submitted":"2020-06-03 16:37:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730953-1591202257689-c70b0bfa","meta_rev_id":"55273b3495aa4ee08684e48ebd6dc6e8d64627a3","_number":730953,"virtual_id_number":730953,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 16:37:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-05-27 12:54:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2020-06-03 15:54:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-06-03 15:54:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 12:54:58.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-06-03 15:54:24.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-06-03 16:37:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26e6ae422e05cb46f4d4ab42c0f4658655cf7d7b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-26 19:52:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ad6a2d0fd9a7fdacf397d3a5c950fffca2c9b9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 20:53:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fb32f83d60d45ac9cb064805664543e : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b61ec746cab456783ffeea51c47a0eb : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c68c60ad2f143dfbb073cfedda0b065 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa163447cea94ff48871e67f6211c28b : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/274d30e5edb546029850ff84d724f33c : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2d748ebf3244f70b7d7417d8776c6c8 : SUCCESS in 7m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/293342f391554ba399d30131ccbb7ffc : SUCCESS in 4m 58s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/edca4b3ad01f4d009ea1f942160b0a79 : SUCCESS in 34m 00s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/8dd1c22672ca40608976bc4bf71012ad : SUCCESS in 39m 18s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/1022c4478a9a43a4b8fae7c06166d561 : FAILURE in 38m 20s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/95722db9d0224d879aeb215580c83db2 : SUCCESS in 45m 55s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/3b486e9225af41849fdc3def1481bdb5 : FAILURE in 39m 57s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/8f24a5fc17c648089c7df52c585a75df : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0e0652520e962578ed34f2bf96de3ae5064d2574","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-27 12:54:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb2aec80df36b5098e5fe32c2347f3769ff97ca0","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-06-03 15:54:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c9e1e0ca53e0c79dc5266073cd68efb9a3c983a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 15:54:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c911206201f7b628d0bbeaef7ff4049e72fab9ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 16:37:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"83c228b55691065aeed18f1bad52d283876a8661","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 16:37:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/692a048c75e4463f97971ece87ee6a6f : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/773d9fd0f370456eb21c7836948a51f3 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1efe7cb3cfdf401f87462aad816b5272 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d4d3ec72bf74dc7a7c3e74b9012da73 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a198666863fd4135a90d07b68a5e6f26 : SUCCESS in 5m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/29c78ddc09e4411fa40f4738b37372ba : SUCCESS in 2m 36s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/85b5ed55ad1745cc9793e89bbdec68f6 : FAILURE in 36m 50s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/0f104ef46ca4440f86be75cebc9c75e6 : SUCCESS in 36m 22s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/2a4a3a8d29594a80a4bec67c927bb53f : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6d9d5c6c67d52e3353d80986baa3d436e596f46f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 16:39:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3486fc53c10e4cc7b84628d4c192411a : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4cf3d16522727edb9a489cbd751664fe0f0df04d","revisions":{"4cf3d16522727edb9a489cbd751664fe0f0df04d":{"kind":"REWORK","_number":1,"created":"2020-05-26 19:52:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/53/730953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/53/730953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/730953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/730953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/730953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/53/730953/1"}}},"commit":{"parents":[{"commit":"4c445c21980bcd9562e84f7cd277f624d357afde","subject":"Merge \"Fix pygments style\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/4c445c21980bcd9562e84f7cd277f624d357afde"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 19:52:07.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 19:52:07.000000000","tz":120},"subject":"Cap jsonschema 3.2.0 as the minimal version","message":"Cap jsonschema 3.2.0 as the minimal version\n\nPrevious versions of jsonschema (\u003c3.2.0) doesn\u0027t support python 3.8 [1].\nPython 3.8 is part of the victoria supported runtimes [2] so we now force\nto use jsonschema version 3.2.0 to avoid issues, remove ambiguity and ensure\nthat everything works with python 3 in general.\n\n[1] https://github.com/Julian/jsonschema/pull/627\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: Ib5308485e50cc076855cc8e1df72b02d208febe1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/4cf3d16522727edb9a489cbd751664fe0f0df04d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/4cf3d16522727edb9a489cbd751664fe0f0df04d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
