)]}'
{"id":"openstack%2Fkuryr-libnetwork~734726","triplet_id":"openstack%2Fkuryr-libnetwork~master~I0adb8922f36365b9c0e3f436341bd03f77b8a5f7","project":"openstack/kuryr-libnetwork","branch":"master","topic":"bp/s","hashtags":[],"change_id":"I0adb8922f36365b9c0e3f436341bd03f77b8a5f7","subject":"Add blueprints link to README","status":"MERGED","created":"2020-06-10 01:03:48.000000000","updated":"2020-06-10 12:08:01.000000000","submitted":"2020-06-10 12:06:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734726-1591790770530-db77c85b","meta_rev_id":"cec95fcd74aa66bcbb2569c7aa05f1d3dfad66e8","_number":734726,"virtual_id_number":734726,"owner":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 12:06:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-06-10 11:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-06-10 04:11:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2020-06-10 11:54:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 04:11:00.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2020-06-10 11:54:41.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-06-10 12:06:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48d9db6ca2b7d7f7f2f2e55f96d680b09fbacc06","author":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"date":"2020-06-10 01:03:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9933452e7cc5f9e3efc3ed9aff282cb170a615bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 01:40:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/75c0ae58dcaf4cdf9a413721eab848e4 : SUCCESS in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c33a5f8a5384a9bad3382fdb756f975 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/263dbacee178426a94ae9e52af1a17ac : SUCCESS in 4m 07s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/c6a3f0b796464aa9a0cafe92b4cee6b0 : SUCCESS in 31m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c44361357fa14253e0994ed68e403d1ab21d927","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-06-10 04:11:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f58983377cd8e8490ab9effa6785226f2c83dcfc","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-06-10 11:54:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea03d75d71073e21b5b678c3fc11cac55d9c66b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 11:54:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa402da33c3786fc9c2dfcc19f270c243d1ee72e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 12:06:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ca1efed2f2b0252cb74a4e8c668b48795f75e434","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 12:06:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b66b285cf0964e31a7d421966ceb1970 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acec07d405bc4a3290dce6acbee07c33 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02124e0817e54aff9878ec6ff15e4b50 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b8d7acee662c2e2b9fa8048886f9bfdcc003b441","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 12:08:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4a651c32a34a5fa23db8c2c31e92a8 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3fc01f57fdddfefc540f11bf5347ea46e3e55095","revisions":{"3fc01f57fdddfefc540f11bf5347ea46e3e55095":{"kind":"REWORK","_number":1,"created":"2020-06-10 01:03:48.000000000","uploader":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"ref":"refs/changes/26/734726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/26/734726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/26/734726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/26/734726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/26/734726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/26/734726/1"}}},"commit":{"parents":[{"commit":"63e7072a89e670f781d8bee3a66a04f94c8c01a4","subject":"drop mock from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/63e7072a89e670f781d8bee3a66a04f94c8c01a4"}]}],"author":{"name":"lidong","email":"lidongbj@inspur.com","date":"2020-06-10 01:01:06.000000000","tz":480},"committer":{"name":"lidong","email":"lidongbj@inspur.com","date":"2020-06-10 01:01:06.000000000","tz":480},"subject":"Add blueprints link to README","message":"Add blueprints link to README\n\nChange-Id: I0adb8922f36365b9c0e3f436341bd03f77b8a5f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3fc01f57fdddfefc540f11bf5347ea46e3e55095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3fc01f57fdddfefc540f11bf5347ea46e3e55095"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
