)]}'
{"id":"openstack%2Fkuryr-libnetwork~738140","triplet_id":"openstack%2Fkuryr-libnetwork~master~I95d3354b4f691850af8cc6f4c6af4c9e98e5cb1f","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"I95d3354b4f691850af8cc6f4c6af4c9e98e5cb1f","subject":"Install tox before running tests","status":"MERGED","created":"2020-06-26 02:54:57.000000000","updated":"2020-06-27 14:50:18.000000000","submitted":"2020-06-26 13:40:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738140-1593178809315-995c8465","meta_rev_id":"9c8f260b9012bb953d0964c5c67269420d107bb4","_number":738140,"virtual_id_number":738140,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 13:40:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-06-26 12:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-06-26 12:52:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 13:40:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d3af16e57c8fa3fd05a0a52a5722ae4a3c81032","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-26 02:54:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34aee3f2a9da0c6ec7696f2e92c4a7c2b1707d3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 03:42:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29dbbd591e414cb49a3589987a6bdfaf : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66dcdda6756d41a5a2ae3c3e11cc5b84 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/107e92f3cd134f739af1f720eaaa412a : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23559ba0f41442cea679255d47728147 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac008c10034f4a9fb34a8fd2633772b9 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6498be221150441ba0103586790d0b78 : SUCCESS in 3m 52s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/0537c3cc465243ab8e75de9b6a230a7f : SUCCESS in 35m 41s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/634bc8b8870a4e72ad4114d99ab5484e : SUCCESS in 40m 28s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/651b7268b400483180342a95efec7995 : FAILURE in 34m 24s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/f63cbc5e1d45433c9ef911581dd77597 : SUCCESS in 37m 15s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/fd5238a3d6ae470cb58e7e14ee80e7bc : FAILURE in 32m 19s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/da944c34cba54ceba12201a5ca743b70 : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":1},{"id":"189aed5166c2cfd249ba353220c4d20b43a0f080","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-26 12:52:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis unblock the gate so fast approved.","accounts_in_message":[],"_revision_number":1},{"id":"7a4845209883471ca0122fe50df1313c260d441a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 12:52:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce26b505cad07f7d22979e604b614ca8cfd0c86a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 13:40:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"45bbe0381ca04b6c4db05302bb178a4cd9e70dd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 13:40:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f80e3a516965440aa755be92fe4cb923 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e7fa95226f34d39a890c61f5a52d86b : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/133be4ac960542699d02a540ccf633d6 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac2cc80d2526462081a5b75414361257 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d242caedd24362acf73cbeda9a1a6b : SUCCESS in 6m 20s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/2245e8fd615945e08e13efcc47250043 : SUCCESS in 40m 55s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/2a6f15d37cb542c19cf6b14afb4c41dd : SUCCESS in 38m 29s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/285dee80921f447b85c29e5757155f60 : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5a5580ce0c0cdb41345ace2e9e2832cff6885d61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 13:41:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007facf9282f4f7f9f43d21f1fa6692d : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cb0e6ba4f7306888d35a7df420911f5afd3ea8c6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-27 14:46:54.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 2f05e8b059f22ad05f74125ea563412e5253b1d6","accounts_in_message":[],"_revision_number":1},{"id":"0663b7b35f2070470d1bae3f323377579616ebaf","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-27 14:49:43.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 7618cafe1fd0123744c45ade1835b4893c20de18","accounts_in_message":[],"_revision_number":1},{"id":"24f6b9c0c61057e2efb3edb873db1f396a278765","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-27 14:49:56.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit edb9a8c293926e46a6fa217d44a9125141e014c9","accounts_in_message":[],"_revision_number":1},{"id":"9c8f260b9012bb953d0964c5c67269420d107bb4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-27 14:50:18.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 8b5a4812f10fd114c1369483403e0ccd3a944121","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"243e500e8cd83edb61972a15db6c6c18a61531b8","revisions":{"243e500e8cd83edb61972a15db6c6c18a61531b8":{"kind":"REWORK","_number":1,"created":"2020-06-26 02:54:57.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/40/738140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/40/738140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/738140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/738140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/738140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/40/738140/1"}}},"commit":{"parents":[{"commit":"47dc0e759fcbf9d92e220a9980a0a49db7b409dc","subject":"Inform that single +2 is required to merge patches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/47dc0e759fcbf9d92e220a9980a0a49db7b409dc"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-26 02:51:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-26 02:54:56.000000000","tz":0},"subject":"Install tox before running tests","message":"Install tox before running tests\n\nWe run tests by using \u0027tox\u0027 so we need to install it.\n\nChange-Id: I95d3354b4f691850af8cc6f4c6af4c9e98e5cb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/243e500e8cd83edb61972a15db6c6c18a61531b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/243e500e8cd83edb61972a15db6c6c18a61531b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
