)]}'
{"id":"openstack%2Fkuryr-libnetwork~738294","triplet_id":"openstack%2Fkuryr-libnetwork~stable%2Ftrain~I3033d28eb268a01de8cf038b1ed20110ca9a31ea","project":"openstack/kuryr-libnetwork","branch":"stable/train","topic":"bug/1881910-stable/train","hashtags":[],"change_id":"I3033d28eb268a01de8cf038b1ed20110ca9a31ea","subject":"Handle subnet without gateway","status":"MERGED","created":"2020-06-27 01:09:54.000000000","updated":"2020-06-29 14:34:47.000000000","submitted":"2020-06-29 14:28:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738294-1593440894366-c88092e0","meta_rev_id":"789b511dbacab047f831a8dde389b549b7b073b8","_number":738294,"virtual_id_number":738294,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-28 18:18:21.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-29 14:28:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2020-06-29 12:54:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-06-29 12:54:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 12:54:02.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-06-29 14:28:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"078e1147a8c816b0eb6ac5ca40fb2a55539053af","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-27 01:09:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0c5b8d72beb46ff6c7c75213db908a51f273b006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 01:47:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de9491168aad46c18999ed2547e01c92 : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3960d898639a4befbc32a2ac231975cc : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc025349e7c04d6da0ed746f7c1e901b : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c928529d4bbf4999a6c81a4f47263c14 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f162682806d24ca8b6e54dc5804181a3 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13bc3f2ed6a649cc8ffe198a50091152 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/436145f3497f4af29a11b6190d43615d : SUCCESS in 4m 00s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/7b4d3e3eb2bf4b5ea20373ab2308aed1 : SUCCESS in 31m 58s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/583a6bea34554e04a0a62c24b9d4f8bb : FAILURE in 32m 00s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/6060c1839941479189e2b70dce29b736 : FAILURE in 31m 50s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/14fb7cf9211247d7b08fd903d7ca3d50 : FAILURE in 20m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0a04bc95047fe6277caf2e872f462b6387362f8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-28 18:18:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5a304fd1d4b95583cd4839e066c30845a86c3298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 19:01:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd0b51ca4cef4a00815a75ada3274f53 : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0c4ea837b714bf89add7362df05e8c6 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b6077945f142d0896371322484e77a : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35dc72aa2d9448bdb1f179a76a497f61 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4604a7573e894082abe2170c5db589ce : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73b1bad3a58544c49592fa958a45e78a : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011ec7f9863c4092afec5449afaab488 : SUCCESS in 5m 48s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/1158f82c207b4930925aa461cb7a88fd : SUCCESS in 27m 00s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/d4781a135f8543b3a5d04eefc1071f24 : SUCCESS in 37m 21s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/485cb126a44f420a8e7b2136681fcd29 : SUCCESS in 33m 44s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/c2a03650895b4c799330b4bfe08c6c84 : FAILURE in 19m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad2f30807194fc3412055a7982615288c5365c8f","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-06-29 12:54:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7dd59bb4c1f3cac03345f606f896b70796c8fbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 12:54:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f171f7f4c83a5188d53836070775c353fc49f452","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 14:28:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0e9f34644744364c47321770d559d7193128ac87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 14:28:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64a1220ae79a4c8cb9ffeacb86a0dd97 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05b35e191d614b5c8adb57dfe63a0735 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/665b6e541928498598d02aa97f48a0ee : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05a5b32d3d3e429a8c3bbe83d0d5d835 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/239608bace76495da33f8f326d1289fe : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/672c55dfdb6e4b4bbeb27eece298f424 : SUCCESS in 4m 51s\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/b60d9ebb48e14cb2a9b5852553d88e62 : SUCCESS in 43m 01s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/22c1c2ffd01641afbec909b75eb9d614 : SUCCESS in 46m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c88a75523c1000498b47b99062f31facb422a131","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 14:34:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb0694e60e2442208adf48ea5fe35a53 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"862c7e73edb44a70d37ae3247c0ba549054b8c60","revisions":{"862c7e73edb44a70d37ae3247c0ba549054b8c60":{"kind":"REWORK","_number":1,"created":"2020-06-27 01:09:54.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/94/738294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/94/738294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/94/738294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/94/738294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/94/738294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/94/738294/1"}}},"commit":{"parents":[{"commit":"3e6bd49035040b32843642b945522a1bb79c692a","subject":"Retry a few times on listing extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3e6bd49035040b32843642b945522a1bb79c692a"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-07 20:43:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-27 01:09:54.000000000","tz":0},"subject":"Handle subnet without gateway","message":"Handle subnet without gateway\n\n* When libnetwork requests the IP address of the gateway\n  (via /IpamDriver.RequestAddress) and the neutron subnet has gateway\n  as None, return \u00270.0.0.0/0\u0027 as a placeholder.\n* Disable default docker gateway via the \u0027DisableGatewayService\u0027 flag\n  (see https://github.com/moby/libnetwork/pull/778).\n\nChange-Id: I3033d28eb268a01de8cf038b1ed20110ca9a31ea\nCloses-Bug: #1881910\n(cherry picked from commit 960993652e1ab526c1ce8cd3cd04d28f13facc86)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/862c7e73edb44a70d37ae3247c0ba549054b8c60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/862c7e73edb44a70d37ae3247c0ba549054b8c60"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
