)]}'
{"id":"openstack%2Fkuryr-libnetwork~738295","triplet_id":"openstack%2Fkuryr-libnetwork~stable%2Fstein~I3033d28eb268a01de8cf038b1ed20110ca9a31ea","project":"openstack/kuryr-libnetwork","branch":"stable/stein","topic":"bug/1881910-stable/stein","hashtags":[],"change_id":"I3033d28eb268a01de8cf038b1ed20110ca9a31ea","subject":"Handle subnet without gateway","status":"MERGED","created":"2020-06-27 01:10:09.000000000","updated":"2020-07-03 06:25:56.000000000","submitted":"2020-07-03 06:23:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738295-1593757390280-f3086ce0","meta_rev_id":"9d081c36d6d55161d4f38b67dac9752e1d7b0b9c","_number":738295,"virtual_id_number":738295,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"date":"2020-07-03 04:56:15.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-03 06:23:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-06-29 12:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2020-06-29 15:37:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 15:37:25.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-07-03 06:23:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c98c32cb6a61b29431613630bd7503082b2eb2f4","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-27 01:10:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c94134d9697759595213424208c04041eb38eacc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 01:44:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07e42b45876940a3b191c0d8c11ba22a : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a711ae61ad340c7a26550cb736d8108 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/401d029a5aef407f8e864762432d52d0 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63c538dfdfd64af59e4e8a1f2b767bae : SUCCESS in 4m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/68c4cd71680d4ee69d293503a058cdd6 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bcbec6c46e84d879779918fd883fc3d : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a244da8418454cedad30ee05170bab40 : SUCCESS in 4m 25s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/337cc396ce8b48e9b3bb608126412cd5 : SUCCESS in 30m 09s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/b58e6bad81664e54aeece8bd817f764e : FAILURE in 33m 13s\n- kuryr-libnetwork-fullstack-py35 https://zuul.opendev.org/t/openstack/build/e1a870015fd64f5e89fdefd9bcf61897 : FAILURE in 32m 09s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/035ab296b82a43bcbda1e036fdbdcedd : FAILURE in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e70e3a8418c72df387721ceb3d62e0f8b972320","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-28 18:18:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f02c4891fa2a75a6c8d274b0c37aff225cef9073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 20:31:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9dfe433835645f9a979eaeb10432aa3 : SUCCESS in 4m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ad15b255724459e932416beb94c85e1 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f049d748a574b10b4bda9fe5d30984f : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/086d8e05d16f46fcbc4985e5ea039b70 : SUCCESS in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/208a3e6dcd9841779b7f638473f38db4 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e49575095f84b3eb7cee053c27203c9 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec785830afee425088a7fc176ff2c36f : SUCCESS in 5m 16s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/502854a4b9344efaa6f9d024725c74f4 : SUCCESS in 30m 01s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/23d99e3673694713befbcb48d6e09553 : FAILURE in 43m 19s\n- kuryr-libnetwork-fullstack-py35 https://zuul.opendev.org/t/openstack/build/285162692e5a4864b3714d8909b5d709 : TIMED_OUT in 2h 11m 39s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/e9135e8bdd234cfba61ac5fc2387d495 : FAILURE in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23af308dab2de2c3e05a6187adb9d604ef3a97da","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-06-29 12:54:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f7c15754f270fc1a425879bcfd3d40dcec08dc7b","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-06-29 12:54:21.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"30d9c71238675aaf9e7838907508375e2d7a80c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 13:42:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9297f6b818d04b588d11e5d276e7422b : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a36c3597b2494da1bd6dff1c1f7d7ce0 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a955ee3cdbad4edc85a5619473ec99e7 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b039307e48e438b8de867b2d6ba3ab4 : SUCCESS in 5m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6e58ea8ef34a4a618a56d6ad43c5fb9c : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71c76adebf9e44068d47e3978e84676e : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dbb801da4da4394a76aed839ddf056f : SUCCESS in 6m 55s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/862ba9e98ec54bd991b1406134c46a39 : SUCCESS in 36m 59s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/ff2a58cf4e614b04901c8c424ff0615d : SUCCESS in 41m 07s\n- kuryr-libnetwork-fullstack-py35 https://zuul.opendev.org/t/openstack/build/0077aaff542e43508ca82ac532989049 : SUCCESS in 35m 17s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/76d02ebf66ca4e4fafc281cfb84a5899 : FAILURE in 21m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"811176f019c222b6396927cedec326fce9a1f2a4","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-06-29 15:37:25.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6b1525de0136d8af229487451567ff6b9af386e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 15:37:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40e66c94fb23295049845870aac7e49bc3f91aa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 16:16:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13ee1690f519467abc94accf40a34fb7 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afec1466d5184a19a29d5ae30d584e0f : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ccc8e92e811418dbc2f4dfe78fa4509 : SUCCESS in 4m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4f31cd3c8e6145a6b00d33748dfb0e27 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22b484d9198f4d79a7ffb39672b76b6e : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f0831c7581949ada82ed5710143f005 : SUCCESS in 4m 41s\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/8b2704a80505488e90bce43e71884f8a : FAILURE in 36m 59s\n- kuryr-libnetwork-fullstack-py35 https://zuul.opendev.org/t/openstack/build/bea33f6672b241db8a87a75f25f95215 : SUCCESS in 33m 11s","accounts_in_message":[],"_revision_number":1},{"id":"15b129aec2041bb6144e33cfc01f6a6e3d22006a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-01 16:26:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"af0c94032d95cfc1611738ad5979ec34cc3e463a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 17:39:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0a48882faa8454c99816760535e2e21 : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1964b9e95374c9890872af1df56a62e : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcc3ced76df846368dfdce2948781dc3 : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4044b8df8b2a4f2f944ca87a00e76da5 : SUCCESS in 5m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f25151880d6c4ab593d4b5c56763380d : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80c25fc3f31e40d7977cbfad225bb366 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/101aa1f80a8f4d49b9f8b980658a845c : SUCCESS in 6m 40s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/1422217fc5614da291f303aeb821781e : SUCCESS in 35m 38s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/05ae13d479c346dc8aae1d660be7c84f : FAILURE in 35m 21s\n- kuryr-libnetwork-fullstack-py35 https://zuul.opendev.org/t/openstack/build/268008942d6d4cb2b91e4bc42ab6f3eb : POST_FAILURE in 1h 06m 01s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/5ab6cf8f9c3a4052a113674729b413ad : FAILURE in 22m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"069245852f69a7b0080acbaefd5d0613b6f7c44c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-03 04:56:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"471f0bbef037428ff673411d0acef92325045ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 05:47:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a67ffdd40e564b338bbc5392b0df2522 : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4497ba2f10d74b0baf17f54c04f3f603 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb35dbf2eb747b0a7a1d4beec602aae : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/955d42bacce54d469ffde6e03fab2efc : SUCCESS in 6m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1712d45723b14cae82d09243b18faf30 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5b047aada18444c9d3e3ff36031dc54 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047dda0dcf9e4684a069deabcc3acde3 : SUCCESS in 5m 47s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/61444428fa9b4b8492125519e282df82 : SUCCESS in 34m 16s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/d6deec4bfc2f4c15bf9cf1b1585e8f54 : SUCCESS in 36m 19s\n- kuryr-libnetwork-fullstack-py35 https://zuul.opendev.org/t/openstack/build/4c155775963942f4a259a39de5f6f748 : SUCCESS in 38m 17s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/5d99c0968d8743b8bf59e84abb8fbaed : FAILURE in 23m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32164ccc2431b5e228ab9ce94c954a729d5e0ed4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 05:47:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"204ba8a6b3b0ba07e4c5e76915822f6445e56c80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 06:23:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f550d3072bb940cdaea8a5bf93ee18bd : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/965f8be34ba04c148c6b7c0abc188760 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/593a2735fe5c40dda3927b37f5b40c97 : SUCCESS in 6m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6734418fe1694831a498227644f62878 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae9a1acda5cd4ba5b6fdcd621936ad1f : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f252e7ee5a14beca9d05192d582ea2a : SUCCESS in 5m 14s\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/024bf7ff48de489f82446eb746c117d8 : SUCCESS in 34m 52s\n- kuryr-libnetwork-fullstack-py35 https://zuul.opendev.org/t/openstack/build/b633bdb9bfed4cb894937b1fa46a7b8c : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0aae738780dfc2ce1b264ccbed03e0468d2f6196","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 06:23:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"317a5803f34fa244c0cf97f779cf55a042ec1874","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 06:25:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfb5ca8172324e3b9dc192e1fa213e38 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8af0405a950e15dd675212b0a7372a547be4b96","revisions":{"c8af0405a950e15dd675212b0a7372a547be4b96":{"kind":"REWORK","_number":1,"created":"2020-06-27 01:10:09.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/95/738295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/95/738295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/95/738295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/95/738295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/95/738295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/95/738295/1"}}},"commit":{"parents":[{"commit":"4682b2dd6456796a72d209bae70d8ececb2c9951","subject":"Update install guide for stein release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/4682b2dd6456796a72d209bae70d8ececb2c9951"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-07 20:43:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-27 01:10:09.000000000","tz":0},"subject":"Handle subnet without gateway","message":"Handle subnet without gateway\n\n* When libnetwork requests the IP address of the gateway\n  (via /IpamDriver.RequestAddress) and the neutron subnet has gateway\n  as None, return \u00270.0.0.0/0\u0027 as a placeholder.\n* Disable default docker gateway via the \u0027DisableGatewayService\u0027 flag\n  (see https://github.com/moby/libnetwork/pull/778).\n\nChange-Id: I3033d28eb268a01de8cf038b1ed20110ca9a31ea\nCloses-Bug: #1881910\n(cherry picked from commit 960993652e1ab526c1ce8cd3cd04d28f13facc86)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/c8af0405a950e15dd675212b0a7372a547be4b96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/c8af0405a950e15dd675212b0a7372a547be4b96"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
