)]}'
{"id":"openstack%2Fkuryr-libnetwork~738312","triplet_id":"openstack%2Fkuryr-libnetwork~stable%2Ftrain~I95d3354b4f691850af8cc6f4c6af4c9e98e5cb1f","project":"openstack/kuryr-libnetwork","branch":"stable/train","hashtags":[],"change_id":"I95d3354b4f691850af8cc6f4c6af4c9e98e5cb1f","subject":"Install tox before running tests","status":"MERGED","created":"2020-06-27 14:49:43.000000000","updated":"2020-06-28 00:52:02.000000000","submitted":"2020-06-28 00:50:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738312-1593305448594-1a3eceda","meta_rev_id":"09e3c06cfd55d648d5cc27dd7efbd8f54f6a99e9","_number":738312,"virtual_id_number":738312,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-28 00:50:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-06-27 22:42:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-06-27 16:46:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-06-27 22:42:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-27 16:46:44.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2020-06-28 00:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e05dd195e5ad6b2398fd7a141c6aa97728e62ffc","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-27 14:49:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cea5925c2bbf37232a8f77e894f633db8c250bf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 15:35:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea1035dd26db45b9885b602199182f5b : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f2a7a0acc2647f68462efd0255e91dd : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1089fc862c954eada338c2ea6b2a5b57 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/737f408ebe7a4d989ec01b932c5edd29 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a6d286d31444f1a8c6e32369b0f39fe : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d5bd397583b44188862974044ca3b56 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d5952bab3354595a6130a13226a9107 : SUCCESS in 4m 03s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/f731b78411f2403d8d4b78244d2706d1 : SUCCESS in 30m 33s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/0bcb7dc66f6e42aabb57731eaeaa7921 : SUCCESS in 44m 24s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/b051e233da8f48d292d5b78a629d52d1 : SUCCESS in 22m 37s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/05b8588056a94effa8becf21f05ea0e6 : FAILURE in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58c233ae072ba95157c41a560cf9bab3efb7a3ae","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-06-27 16:46:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"98a30a5721f192e4041fbe466ac3995da158dca9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-27 22:42:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nFast approve this patch for unblocking the CI","accounts_in_message":[],"_revision_number":1},{"id":"3d40182ce5d777c72a7512858b828d1a1005eb77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 22:42:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0478984903933d3c8429f5529b4fc2c66134ce78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 23:27:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d580ea5f26c42e8b2ef5edac98c5a01 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b617fcf10fa4458bae3d1e8b81b275ae : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e005b56077a3454a8a38a5615fa8188d : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22c2b51a97a84b86bd7ca0fc69488f6b : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c297e474cd44a9991560a878db5af6e : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c40c8649c88436ebe44e2658751354d : SUCCESS in 4m 59s\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/b10fd0fd522442e4a321a5cfe8e97a8d : FAILURE in 43m 46s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/6b9eb28aa8684e81b5cb181c77b37c12 : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"809f2f9b4d463c983f00305c0ff4d3e82477147d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-27 23:42:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"05435728344787d1ca81fc5402348e31662287f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 00:16:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7de7b82a5faf4d21926c103975c06199 : SUCCESS in 28m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/faa3e195bc694b419c51190fcb101553 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cae0b65d0304c889255d3272eca5276 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67a26e3a0182465ebd3f69f996f76976 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53274e4375284f9dbd580213fe593a2e : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/033b33262d1244ea8fe82ba42304d2ed : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/064c3a000dd84b2c8cb7dd6db7793c8f : SUCCESS in 4m 28s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/882721364b4e40b4996305cf684e8930 : SUCCESS in 31m 55s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/c522d522fc9f4b5a889dad368a7f1f6e : SUCCESS in 32m 55s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/dfa2c52cbcfb4dbd82acf87d725a4775 : SUCCESS in 30m 43s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/c22b2226e4904e519b3f12df376f0efe : FAILURE in 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb2acc788692144ff42bd20b3b1cd89b831efb96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 00:16:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a2591a49f840ad434ec4f7570f7bb5c5e1e76fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 00:50:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3208a588ddbc42c5ae643feead3d1c5c : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a9d6428b3a942f0af3cd23c238c65ae : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b5ff96c78d9423a87474c5bddf88cd3 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51db4405d7aa47a699859f083ce85dc0 : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a31504f40f84684a88b78a55a20e0e9 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7571d7178177462b8adab8a6c1918d95 : SUCCESS in 6m 01s\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/4d2ff09e256544bdb36e5dfa7f3ffefa : SUCCESS in 32m 31s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/03104ac4b10c40d0abd752a113e96a58 : SUCCESS in 32m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0bfade614ea44077d491286430b4e15a407208a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 00:50:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"09e3c06cfd55d648d5cc27dd7efbd8f54f6a99e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 00:52:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09fb03f05ce4473fac8989a92d4d6d16 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7618cafe1fd0123744c45ade1835b4893c20de18","revisions":{"7618cafe1fd0123744c45ade1835b4893c20de18":{"kind":"REWORK","_number":1,"created":"2020-06-27 14:49:43.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/12/738312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/12/738312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/12/738312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/12/738312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/12/738312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/12/738312/1"}}},"commit":{"parents":[{"commit":"3e6bd49035040b32843642b945522a1bb79c692a","subject":"Retry a few times on listing extensions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3e6bd49035040b32843642b945522a1bb79c692a"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-26 02:51:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-27 14:49:43.000000000","tz":0},"subject":"Install tox before running tests","message":"Install tox before running tests\n\nWe run tests by using \u0027tox\u0027 so we need to install it.\n\nChange-Id: I95d3354b4f691850af8cc6f4c6af4c9e98e5cb1f\n(cherry picked from commit 243e500e8cd83edb61972a15db6c6c18a61531b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/7618cafe1fd0123744c45ade1835b4893c20de18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/7618cafe1fd0123744c45ade1835b4893c20de18"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
