)]}'
{"id":"openstack%2Fkuryr-libnetwork~738314","triplet_id":"openstack%2Fkuryr-libnetwork~stable%2Frocky~I95d3354b4f691850af8cc6f4c6af4c9e98e5cb1f","project":"openstack/kuryr-libnetwork","branch":"stable/rocky","hashtags":[],"change_id":"I95d3354b4f691850af8cc6f4c6af4c9e98e5cb1f","subject":"Install tox before running tests","status":"MERGED","created":"2020-06-27 14:50:18.000000000","updated":"2020-07-04 23:10:51.000000000","submitted":"2020-07-04 23:10:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738314-1593904251148-fcaafab0","meta_rev_id":"74c92cd105b908e4bc769b5f8c563ebddcdb69c6","_number":738314,"virtual_id_number":738314,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-04 23:10:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-07-03 12:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-06-27 16:47:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-07-03 12:37:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-27 16:47:03.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2020-07-04 23:10:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"002b5e4bd56bf4775057c3918477133e1a90f74f","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-06-27 14:50:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2217a8f5b1a21a465258f3d0d76c4d63824f2554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 15:08:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17fe3c240d714b2e860ce88c744faebb : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/94133af0dc8d4e1796d94f2061b7643b : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5670e1bd682d4855a754d6d8bc7a870f : SUCCESS in 3m 54s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bbcc07fd989346fa84fca4ecb279c955 : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2591420dcb6b4df7a8e69ee90ce0ea16 : SUCCESS in 6m 00s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/398ad613f6b645fc82056d922e018437 : FAILURE in 10m 32s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/43239f642eae47d8beb073772da5894a : FAILURE in 12m 16s\n- kuryr-libnetwork-fullstack-py35 https://zuul.opendev.org/t/openstack/build/26971706f27d4e469902bcdb5ea47cf7 : FAILURE in 17m 13s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/2723c1bad0424818a52c921d99c30474 : FAILURE in 8m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4adfbf46b7f14842927a83479a7dc947 : SUCCESS in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a383e878a4e10b079b84acb9a01d585beb9593ba","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-06-27 16:47:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d7a30243dcb37d13582217c625e4d23b4da819ef","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-03 04:56:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ed7f6cbe2a564726f42ec70ab842a2f59a4eb86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 05:48:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ca9af2eefce499499a7251e0bfa2374 : SUCCESS in 7m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f114f46a47de434b842124de10730072 : SUCCESS in 7m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3aa0065cf82c422aa726856d8e3c02aa : SUCCESS in 6m 36s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2f0fadc7e59b4eb8a097449bbffc24a2 : SUCCESS in 7m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/151988ef294b47aa9969af9553314b91 : SUCCESS in 8m 25s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/c974fd2c9c89476ca070ba7763cef76b : SUCCESS in 27m 09s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/d4103c0a99474965ad93932bddbd2a86 : SUCCESS in 34m 20s\n- kuryr-libnetwork-fullstack-py35 https://zuul.opendev.org/t/openstack/build/1dd90145ffb24fe880842fc69418e0da : SUCCESS in 42m 24s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/64db4cf5059d48e9880b755bf023df1e : FAILURE in 21m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/639cff68122a4c688fac7d4a32c4bd13 : SUCCESS in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5cfd62ff75095d4b90c625f3f0946c8da4857900","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-03 12:37:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis patch unblock the CI. Fast approve.","accounts_in_message":[],"_revision_number":1},{"id":"3fe7d6733383cb52a712a32d8510b49b960e3036","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 12:37:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eac7599dd3c7a2378ff76b18bdc12da665a14617","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 13:23:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/584a1d5738a444749149a4c1ac1ae062 : SUCCESS in 7m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f3e19208ce54e1a85f71f64abd588cd : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/90e93b0a07e74c43b00809898a54ea7f : SUCCESS in 5m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/49eb6f06d94649949343959ec3ddf10c : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/510215f9a6624706ae6e82075570b5c2 : SUCCESS in 8m 08s\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/a29c73e851f743ab832f3251a95683b8 : FAILURE in 39m 18s\n- kuryr-libnetwork-fullstack-py35 https://zuul.opendev.org/t/openstack/build/03ace5925fec499bb6227657175220ac : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2ae93c0f881548f9398cbd870177d98f43d7b999","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-04 21:53:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"645591f14c135f3291aa9421e05b989d0ff90091","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 22:36:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5e577cd031546f990efbab03f022fdb : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f7001ff78ea453ab2327da9f9f1120d : SUCCESS in 4m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4d0d0b9075f04dff897d8dda14a38004 : SUCCESS in 4m 33s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5aa6199ed36544f484bb6f6ce11a958f : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/598360665b0b4ec6b46cc9f2ef52b74a : SUCCESS in 4m 39s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/7e5ef0a218b047cfb381c64a2da459f4 : SUCCESS in 33m 02s (non-voting)\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/f96708f89fcb419fb5f1d42d02d1117c : SUCCESS in 41m 00s\n- kuryr-libnetwork-fullstack-py35 https://zuul.opendev.org/t/openstack/build/60acac96e91d4e24ae44c87b896bfea1 : SUCCESS in 35m 46s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/5da0a1d5ec414e1ab1f2668b6fb425b4 : FAILURE in 17m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32205d7b45ef4dc79d1d3e82340817d3 : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8344e99a16a33b4bb34477edc9b7a50c975293d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 22:36:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab16f6af81015d34691c751074582cec5c29a0e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 23:10:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9997cbc066a4af1917f5ccd86104027 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6be738e98f2d42658ba2c2c0d3f3823a : SUCCESS in 6m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/520c0e8fe1e74989bd8f1fe3e8f88d81 : SUCCESS in 4m 29s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/efa39a8840e54903afa058081a4db785 : SUCCESS in 4m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5881dcc6b820423c9e644c489f5c67d0 : SUCCESS in 4m 21s\n- kuryr-libnetwork-fullstack https://zuul.opendev.org/t/openstack/build/ea8fb235b8994e88abd62c31332d2f92 : SUCCESS in 32m 19s\n- kuryr-libnetwork-fullstack-py35 https://zuul.opendev.org/t/openstack/build/85883d36af4c4ceb8e981b38031cfe79 : SUCCESS in 26m 08s","accounts_in_message":[],"_revision_number":1},{"id":"74c92cd105b908e4bc769b5f8c563ebddcdb69c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 23:10:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b5a4812f10fd114c1369483403e0ccd3a944121","revisions":{"8b5a4812f10fd114c1369483403e0ccd3a944121":{"kind":"REWORK","_number":1,"created":"2020-06-27 14:50:18.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/14/738314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/14/738314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/14/738314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/14/738314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/14/738314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/14/738314/1"}}},"commit":{"parents":[{"commit":"da2b3fd4ce18a907ad409794d34242fdf8200dad","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/da2b3fd4ce18a907ad409794d34242fdf8200dad"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-26 02:51:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-06-27 14:50:18.000000000","tz":0},"subject":"Install tox before running tests","message":"Install tox before running tests\n\nWe run tests by using \u0027tox\u0027 so we need to install it.\n\nChange-Id: I95d3354b4f691850af8cc6f4c6af4c9e98e5cb1f\n(cherry picked from commit 243e500e8cd83edb61972a15db6c6c18a61531b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8b5a4812f10fd114c1369483403e0ccd3a944121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8b5a4812f10fd114c1369483403e0ccd3a944121"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
