)]}'
{"id":"openstack%2Fkuryr-libnetwork~741832","triplet_id":"openstack%2Fkuryr-libnetwork~master~I54d784fe0bded5f461ba9f26ed901ec042a6bd0d","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"I54d784fe0bded5f461ba9f26ed901ec042a6bd0d","subject":"Migrate jobs to zuulv3 native","status":"MERGED","created":"2020-07-19 03:52:09.000000000","updated":"2020-07-26 00:41:31.000000000","submitted":"2020-07-24 03:28:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741832-1595561337317-0ec4c45c","meta_rev_id":"8fc1c28407829046d7da1ee4b78ad834bbde1daf","_number":741832,"virtual_id_number":741832,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 03:28:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-07-22 13:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-23 14:27:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-07-24 02:48:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 08:34:46.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-07-22 13:57:48.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-07-23 14:27:29.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2020-07-24 03:28:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5dae60502a3d77e0163c1e773081f00b9b2fd8a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-19 03:52:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2127412f36d6162220af1c7eef39a759f0e70a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 03:52:24.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kuryr-libnetwork on branch master.  The error was:\n\n  Unknown projects: openstack-dev/devstack\n\nThe error appears in the following job stanza:\n\n  job:\n      name: kuryr-libnetwork-fullstack-base\n      parent: devstack-minimal\n      #vars:\n      #  use_python3: 0\n      #  use_src: 0\n      #run: playbooks/kuryr-libnetwork-fullstack-base/run.yaml\n      #post-run: playbooks/kuryr-libnetwork-fullstack-base/post.yaml\n      timeout: 7800\n      required-projects:\n        - openstack-dev/devstack\n        - openstack/devstack-plugin-container\n        - openstack/kuryr\n        - openstack/kuryr-libnetwork\n      roles:\n        - zuul: openstack-dev/devstack\n      irrelevant-files:\n        - ^.*\\.rst$\n        - ^doc/.*$\n      vars:\n        use_python3: 0\n        use_src: 0\n        devstack_services:\n          # Ignore any default set by devstack. Emit a \"disable_all_services\".\n          base: false\n        devstack_localrc:\n          KURYR_CONFIG_DIR: /etc/kuryr-libnetwork\n        tox_envlist: fullstack\n        tox_install_siblings: false\n        #zuul_copy_output:\n        #  # We need to copy directory with logs to have it in job artifacts also,\n        #  # /opt/stack/logs is default logs directory defined in neutron\u0027s\n        #  # tox.ini file\n        #  \u0027{{ devstack_base_dir }}/logs/dsvm-functional-logs\u0027: logs\n        devstack_plugins:\n          devstack-plugin-container: https://git.openstack.org/openstack/devstack-plugin-container\n\n  in \"openstack/kuryr-libnetwork/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"87db4cebb531cd557677811e3365a13c3f1abdf8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-19 03:57:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"58ba19e0039084413807af93660499e4c475bfc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 04:31:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c51a385f25943d48fdcd69bf0b79cea : SUCCESS in 8m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9c55341121b4b48849b67bed64fb823 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa939ef93e2a41a5adbb291fbad12ccc : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72973e343ac543dab615d653edaa4744 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ffd280b191845ebb31544cf5bec553b : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34fe259721b048a4a2b713bb713a85d6 : SUCCESS in 4m 00s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/e2ce216483954ddc9f9cb31b11952d7d : SUCCESS in 8m 48s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/3798181a063c4502984576c7cd67382d : SUCCESS in 8m 16s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/b704c7a6206a4af9a310ab70278d36b8 : FAILURE in 30m 10s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/ced3b9ddabdd46fa86201286c72dcf4a : SUCCESS in 8m 12s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/be1f8cd8e5d94631a1c16b7a66374a15 : FAILURE in 32m 36s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/29d74022db564673a2ac4a79c37dcb26 : FAILURE in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f52c0780210ffd00ef787b45cce2517677c65c2b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-19 19:02:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4ddead9c7d7f9fb8854d2e22c19135ed538a33ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 19:43:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18cacb5ffc914b22a8098f9eb8d2e968 : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/157a01bda64948b791a250d0965797e4 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/361af015fde54f30a11169224a24ad00 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/804c94ccff5748e5a807612a9ddb24c3 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13b186845804439e86a7531d2d6e7df8 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b1702695cf94b0780b1049ecfa46188 : SUCCESS in 3m 28s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/59469d23b5374769ba7a3338b0cde16e : SUCCESS in 13m 14s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/874fe6802ea941faa0a926f4063710f1 : POST_FAILURE in 14m 28s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/69cbff2129ff4aad9d042c8232e84b57 : FAILURE in 34m 49s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/5ef4bec6bb44445bbefb7b0983367e49 : POST_FAILURE in 14m 22s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/0e89490875fb4b85b9d9f2c8f4c9533d : FAILURE in 34m 51s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/7d20e15cac594747bc418ab8cc7de993 : POST_FAILURE in 9m 51s","accounts_in_message":[],"_revision_number":3},{"id":"57b98955c3cb4c8c57705776228eb8c3b5360e1d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-19 19:55:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"beb01d00783a1401e432a791e30e270c0f519cf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 20:36:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a391eab2597e4fbf833dbeb1f96342a2 : SUCCESS in 10m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00c3acd5470f49e7bed67ae733cb0366 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c82603324674456a897b9be4e1a7489 : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd26de153583483884ede156a57edbac : SUCCESS in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/636d7d64f0e34934a517a9dc5154c219 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da8b8f4bbe340448dbbd08b08db6cf8 : SUCCESS in 3m 32s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/bfedf0214d6b45b6b1fcfa259a14db01 : SUCCESS in 11m 06s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/d5a392a843a04cccb9381411eee3d739 : POST_FAILURE in 24m 06s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/53e03e08a2a04cac809301271cdec2ae : FAILURE in 35m 00s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/1a1e9309ca714addb0d119df10fa666f : POST_FAILURE in 13m 12s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/906d78a7c9bc4ed48603c3314e87c06b : FAILURE in 27m 04s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/2286a54b6bd347f2a75bb87b1d0b04de : POST_FAILURE in 7m 59s","accounts_in_message":[],"_revision_number":4},{"id":"eece438f38adc3b311d76017f28bb486fae4331e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-19 20:49:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cabe42c0bd2395f78a4f3a58ac4c26390add132c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-19 20:51:50.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a678d4858f22d6a14e85cf6de400cfff4aa3db95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 21:31:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e361257d0de4a2c9edbe973bb1d6213 : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a970c05de0c84c7994200d2937b92e75 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8f9a02971c74f8c90fff6899eacbf1b : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a79e5df09134683b1fc599422152bbb : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7a5ccb8cdab41c183e1b142f1fe932e : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dfa394e2370481cbc38a5cde82d2564 : SUCCESS in 3m 48s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/d51fbde1f5134e4193b50ad81fbe39c2 : FAILURE in 13m 37s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/eaf737ccffd5473dbc51e53d5d7867d9 : POST_FAILURE in 13m 38s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/b747030d6e0c420a8ade50756378c6ea : FAILURE in 32m 33s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/990773c96bf543528515e271552f382e : POST_FAILURE in 14m 00s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/dda787dcb495448398d8f81ab2ff7078 : FAILURE in 32m 33s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/d4b579a7af8741cc81b610fba35b2373 : POST_FAILURE in 33m 53s","accounts_in_message":[],"_revision_number":6},{"id":"90c1ed2afa02ce426679178582cfffbbe2b7b13a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-19 22:13:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5b56eb0cc28e8ece8e4e1dd4143c5d86e1b9e1f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 22:42:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3db398cb63840079ca22f591fca8963 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc36129ce98b4545abcd2b4187f1edb9 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/949bc16a80fa47b2ba56a5cf865fdb40 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb0e9f2938eb4e06873d2a435d6dd684 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9501233b65c944779a05462f4099a7ae : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9534f8f00a5143388a277120ce3be1a6 : SUCCESS in 4m 15s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/df2986af18c9430d9924a9c13072b8dc : FAILURE in 9m 49s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/9477a0b1a97646d5a817351856a72bcd : POST_FAILURE in 26m 16s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/14bb06a45bb64ec89e72d9511dbfef50 : FAILURE in 27m 34s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/2a71a14960044643981f7400e4a41bac : POST_FAILURE in 26m 24s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/145351b7a2cc49f3b5cf4cf4f688f000 : FAILURE in 27m 39s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/6510d1c57de94edf9d24f2a74359efa9 : POST_FAILURE in 26m 45s","accounts_in_message":[],"_revision_number":7},{"id":"b6d43573b7dcad5afe7f5053a5d4974f2ab31e52","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-19 23:02:25.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6a0f355bde862013c4eee630d8942869dbbb5a96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 23:41:38.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4ba55ea3bc0439e90957d8e1a749f17 : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51a7c9c055e14fbca201cceb3c4ae56c : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc0f857c21b64e38adea7a7822d0bf6b : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72fdc10edc754b69a8c4519197f03668 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4c7db3e88fc4a48833debb9fba8ba6d : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8703157dd23c496998400cac388f6183 : SUCCESS in 3m 51s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/ccdf8eacbd2c46cbb915b75a0cdabe62 : SUCCESS in 27m 54s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/e2a5dc279a4e4d658bb148a62598e7c2 : SUCCESS in 30m 11s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/68459c1a926842edb65c60d01b430a04 : FAILURE in 31m 07s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/54b1f329fe1b4baab98b22c8de39e231 : FAILURE in 28m 59s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/d046f2febd7b4735aecfcb487001db97 : FAILURE in 33m 08s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/11a830aede9d4036b97224f2c9c0b5b3 : SUCCESS in 30m 35s","accounts_in_message":[],"_revision_number":8},{"id":"cf5d3196d63c3d155a729b2953defb273387fc77","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-20 00:10:44.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"e57700a60266e2ffb733ffd368b8fd47f835e81c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 00:47:16.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fee25e5b00a46d1b0a2c82edcb5832e : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7de57d76b5a4acda8b193d7c622e2d9 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d914376571400594cd4067a9fe36a9 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbc81f6ba0214089805d7c6e2bea3277 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a536c9c1e1cf414abdcf5755ced86a9b : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea7c780560b54fb7adf7f1c60ce4fd32 : SUCCESS in 3m 36s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/dab938f77388437cb2ccee453ff359dc : SUCCESS in 27m 40s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/b6fb0fe31bb24b289d422a4c8156a14c : SUCCESS in 28m 45s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/15967c485336495d949f923818d7c306 : FAILURE in 29m 18s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/c9475ce54a3f4e86b964c42d467a6511 : SUCCESS in 31m 13s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/bb3c4871e0514659b7f20cf9d7681f8d : FAILURE in 27m 30s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/9aa7c4d0a68543259d1315fb0468401f : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":8},{"id":"6a0104046a4136d4d26fd5caf3ab42eafbee2482","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-20 01:27:27.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"16ec140f337216f8d00a610a7a8ef56f240040d7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-20 01:31:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d0d792c746db9255fafb91776689d944c4973888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 02:09:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c683d80ba6fd454c82e93afb096d3bbf : SUCCESS in 3m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5f3c5e1ca9c42e3915e8ba9ef0d5f22 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c627bb6ac279448d8ef9f622ca004c43 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/726286508bdf479cb5667754214eed2f : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f9da421e5a14a0683af2d90cabe251e : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c89e58e57d3488aace170a9e329e8dd : SUCCESS in 3m 17s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/b73ef833da03473abdc0d25d0d2b0f56 : SUCCESS in 29m 24s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/72bb6e3d73ff414da5e4f1291ff5a908 : SUCCESS in 30m 15s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/e96bfc3615c6487098a863a1ac0444d4 : FAILURE in 22m 11s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/59276951c1f94039a29686bd932f481e : SUCCESS in 31m 19s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/c848fdb6964e4acd95d1a9b5fad4ee7d : FAILURE in 32m 17s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/c6ee700494924d92a5ea22540dcb951e : SUCCESS in 21m 44s","accounts_in_message":[],"_revision_number":10},{"id":"4054cad11927d95c1a5575e1dfbd61ef05343230","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-20 02:33:16.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2beb1b55279fea7ed2bfbadcf2e847debf06ccf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 03:06:31.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28671efef9cd4e43857024863f91993b : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/175f0de9c33d4eae9406b6076b01614b : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/734d96f5bf504372b9ecf40c7e75a506 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9b767ed5f7e406bbd708a0fa6184d36 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a5f3b61cdf14ffd9e462cb626110f23 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/910c88647d2849b7b4721210ab6a4f76 : SUCCESS in 4m 16s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/c1abe8edace6480caa3447dbca22f90a : SUCCESS in 26m 42s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/a7c46722fcae40a9a5d4544535537bed : SUCCESS in 30m 36s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/80c7554bff634596897343f1b3c374ed : FAILURE in 31m 41s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/a3f0d60d6fac495eba3889e91d92c720 : SUCCESS in 30m 22s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/729bb333483f4fed8981238d63062a01 : FAILURE in 29m 29s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/d88572c8e36f4f038f3819a1fb89260c : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":11},{"id":"a412c1eb75452aae9c55b4fbe90f7c56b0c12e7a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-20 08:34:46.000000000","message":"Patch Set 11:\n\n(4 comments)\n\nThanks for working on this! Please use the gerrit topic native-zuulv3-migration","accounts_in_message":[],"_revision_number":11},{"id":"98355d53b3e415471425d7b44cef093a81921433","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-21 08:07:08.000000000","message":"Patch Set 11: Code-Review+1\n\nLuigi\u0027s comments make sense, but otherwise it looks okay to me.","accounts_in_message":[],"_revision_number":11},{"id":"4614c55931c739695a044387047015795af8dd30","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-22 02:06:54.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"84be4450575bbbd980dfa587abf6074baaf404c8","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-22 02:08:31.000000000","message":"Patch Set 11:\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c9b422c7cb38a03ba24182596b423f257a023b32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 02:51:43.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80cc319412a945ea874aa4926e5f106f : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5a68f3ee2194aa5b37a16dbb491db52 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f0c4fa3a3544cc3bf380cf4d986842b : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58d1c582082a41fdbe4f8db9bfcc5a6b : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e351003a422478180deb9de256f0dfe : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153200d4af6d46f9bbba2f10b8c00c18 : SUCCESS in 6m 33s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/282b4e59d4ec41679fe6ceefeceb39e4 : SUCCESS in 31m 40s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/89208041f1a341578580e6190000e44a : SUCCESS in 23m 11s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/7c9432dbe29b4fcd97fb2770c2a61b36 : FAILURE in 35m 01s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/f09d4b54e141482b88f51acf967f29c2 : FAILURE in 42m 07s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/92e2a9e974184dc09fdd5d8d610c0e54 : FAILURE in 28m 36s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/a737816b1f1b4d41beeb8ce0fffbc5ba : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":12},{"id":"7bc6f92dad1c54a233400dc5f82722823b3e69ec","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-22 03:36:42.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"296a838bd16cc7a1aae8faad114ee727ff85ae0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:29:28.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0573f6e1a07d49449d93365438364d60 : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5d1412ed094452db136155e20047104 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975d333fc0ee472e8681597ab0397e5e : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cbc9b9d7107464d8498240cef43d7b7 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb9a7ca292c646018d47932fae8b5581 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1ffd21661e40c0a60dd5572e9932e2 : SUCCESS in 3m 52s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/003dc42448f0416b90ba9f57319bf147 : SUCCESS in 20m 35s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/349c15d98acb4ea397cd660aba436363 : SUCCESS in 22m 33s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/6ca3112760664c4aa7cc56a47f2704ec : FAILURE in 22m 22s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/db6758359bd04548a0c4eb552c9e2e12 : SUCCESS in 29m 48s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/e8763d98e6724f6d829b9483e272a60e : FAILURE in 22m 54s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/b43a7f5be4d14ec682833bb4becd07c6 : FAILURE in 35m 09s","accounts_in_message":[],"_revision_number":12},{"id":"2185dd3137b1fde2734d6c7242d270a0304d4669","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-22 13:44:47.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"6fd7dcf0ee900f2e63e02b8c965d1b08ecf96556","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-22 13:57:48.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"57f0b499df080a1f98d632c4c0153ff067a7699d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 15:06:32.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c920513c2b7648b0a9ea07b6198bcd65 : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f3f73bdc3324fef90a91af750db1e50 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d7696eff414d0ba33d189f66939cf5 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a27fc5815df2445d847260cca85c49e1 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01a07045370e4f099991f33832540c08 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d995eaa3064475bac42f15d844d88eb : SUCCESS in 5m 00s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/d98b359de0524015a0acb23c23da1ee8 : SUCCESS in 32m 00s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/3be304c4ff1e43aaa228a06da5711b08 : SUCCESS in 39m 58s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/51b2a22c7f76403bb86f456565caedbe : FAILURE in 33m 49s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/81b5d6619c2a453993acdc228832e5d2 : SUCCESS in 35m 12s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/0017950f1ce340dda93ea78766578dc3 : FAILURE in 50m 22s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/f4843c650f3c4924a108cf64f39790e9 : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":12},{"id":"0ff03477542abb8c4b8f3bb35c43f7ef9fda14d9","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-07-23 14:27:29.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"61048028ab9cb3fa390feeccbd69da8cd5790a61","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-24 02:48:39.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"45e91b622511aa7e8312b4ec2af9720854cf09f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 02:49:00.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"f7d6f47736daadd9c69542f48639cba09f78b25f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 03:28:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"efc2442f272918e9d2804bfbaa7c9b78c54c42e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 03:28:57.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/310b368bb497425dbb8dfcb077c5a46b : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24c0bb0d2f8b401997c52ce707abab19 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89380954c7a242aeb7ddcd8b45f65144 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aeb1aa03d64d4473a4f5e0964ed1691a : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf12824e0c24b8c84db63bc6bb9b875 : SUCCESS in 4m 41s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/2c92fb4766544a38b97ca99e77f10a93 : SUCCESS in 31m 44s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/0852d3650f594fb9960afe2baec8dc3c : SUCCESS in 32m 56s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/4dd798c52bf14159b98baf2ab2297a54 : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":12},{"id":"a829f374b9858e20641dbcb077bb16d516593b46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 03:33:39.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99b1085a21c04f499c0149b798596112 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":12},{"id":"8bb8c521df3512b9edbcee71666f95da9cd900b5","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-24 07:58:07.000000000","message":"Patch Set 12:\n\n(1 comment)\n\nThanks! A bit late, just a comment:","accounts_in_message":[],"_revision_number":12},{"id":"8fc1c28407829046d7da1ee4b78ad834bbde1daf","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-26 00:41:31.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"055745136e5715f0f5bb1828cbcc46cf9be145ba","revisions":{"d3bbbc40c8458da86e050f1487d1e2b49e8f22f9":{"kind":"REWORK","_number":1,"created":"2020-07-19 03:52:09.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/32/741832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/32/741832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/1"}}},"commit":{"parents":[{"commit":"3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7","subject":"Merge \"Handle subnet without gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:50:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:52:09.000000000","tz":0},"subject":"[WIP] Migrate jobs to zuul v3 syntax","message":"[WIP] Migrate jobs to zuul v3 syntax\n\nChange-Id: I54d784fe0bded5f461ba9f26ed901ec042a6bd0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/d3bbbc40c8458da86e050f1487d1e2b49e8f22f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/d3bbbc40c8458da86e050f1487d1e2b49e8f22f9"}]},"branch":"refs/heads/master"},"a66ac1f7a575285d1d1fda9fc7f8778f29eded5b":{"kind":"REWORK","_number":2,"created":"2020-07-19 03:57:42.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/32/741832/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/32/741832/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/2"}}},"commit":{"parents":[{"commit":"3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7","subject":"Merge \"Handle subnet without gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:50:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:57:36.000000000","tz":0},"subject":"[WIP] Migrate jobs to zuul v3 syntax","message":"[WIP] Migrate jobs to zuul v3 syntax\n\nChange-Id: I54d784fe0bded5f461ba9f26ed901ec042a6bd0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/a66ac1f7a575285d1d1fda9fc7f8778f29eded5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/a66ac1f7a575285d1d1fda9fc7f8778f29eded5b"}]},"branch":"refs/heads/master"},"b2352702d5dfccf46248b01ea2e1104f032a7e21":{"kind":"REWORK","_number":3,"created":"2020-07-19 19:02:54.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/32/741832/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/32/741832/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/3"}}},"commit":{"parents":[{"commit":"3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7","subject":"Merge \"Handle subnet without gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:50:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 19:02:48.000000000","tz":0},"subject":"[WIP] Migrate jobs to zuul v3 syntax","message":"[WIP] Migrate jobs to zuul v3 syntax\n\nChange-Id: I54d784fe0bded5f461ba9f26ed901ec042a6bd0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/b2352702d5dfccf46248b01ea2e1104f032a7e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/b2352702d5dfccf46248b01ea2e1104f032a7e21"}]},"branch":"refs/heads/master"},"7e3c7efe16193067bb9c28fee6fc5f5244aac255":{"kind":"REWORK","_number":4,"created":"2020-07-19 19:55:10.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/32/741832/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/32/741832/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/4"}}},"commit":{"parents":[{"commit":"3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7","subject":"Merge \"Handle subnet without gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:50:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 19:55:05.000000000","tz":0},"subject":"[WIP] Migrate jobs to zuul v3 syntax","message":"[WIP] Migrate jobs to zuul v3 syntax\n\nChange-Id: I54d784fe0bded5f461ba9f26ed901ec042a6bd0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/7e3c7efe16193067bb9c28fee6fc5f5244aac255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/7e3c7efe16193067bb9c28fee6fc5f5244aac255"}]},"branch":"refs/heads/master"},"be2cc723a1ffe30794c5dedeb22578cd5465b073":{"kind":"REWORK","_number":5,"created":"2020-07-19 20:49:06.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/32/741832/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/32/741832/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/5"}}},"commit":{"parents":[{"commit":"3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7","subject":"Merge \"Handle subnet without gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:50:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 20:49:01.000000000","tz":0},"subject":"[WIP] Migrate jobs to zuul v3 syntax","message":"[WIP] Migrate jobs to zuul v3 syntax\n\nChange-Id: I54d784fe0bded5f461ba9f26ed901ec042a6bd0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/be2cc723a1ffe30794c5dedeb22578cd5465b073"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/be2cc723a1ffe30794c5dedeb22578cd5465b073"}]},"branch":"refs/heads/master"},"9d23022b3b55bd3a818eb01e9cfd593ed2db9c93":{"kind":"REWORK","_number":6,"created":"2020-07-19 20:51:50.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/32/741832/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/32/741832/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/6"}}},"commit":{"parents":[{"commit":"3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7","subject":"Merge \"Handle subnet without gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:50:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 20:51:44.000000000","tz":0},"subject":"[WIP] Migrate jobs to zuul v3 syntax","message":"[WIP] Migrate jobs to zuul v3 syntax\n\nChange-Id: I54d784fe0bded5f461ba9f26ed901ec042a6bd0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/9d23022b3b55bd3a818eb01e9cfd593ed2db9c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/9d23022b3b55bd3a818eb01e9cfd593ed2db9c93"}]},"branch":"refs/heads/master"},"973aae77675a344fd89eb1ae1b8cca3fd2234136":{"kind":"REWORK","_number":7,"created":"2020-07-19 22:13:05.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/32/741832/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/32/741832/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/7"}}},"commit":{"parents":[{"commit":"3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7","subject":"Merge \"Handle subnet without gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:50:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 22:13:00.000000000","tz":0},"subject":"[WIP] Migrate jobs to zuul v3 syntax","message":"[WIP] Migrate jobs to zuul v3 syntax\n\nChange-Id: I54d784fe0bded5f461ba9f26ed901ec042a6bd0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/973aae77675a344fd89eb1ae1b8cca3fd2234136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/973aae77675a344fd89eb1ae1b8cca3fd2234136"}]},"branch":"refs/heads/master"},"01cfd8fc208aec252c11020304a57935b8df319c":{"kind":"REWORK","_number":8,"created":"2020-07-19 23:02:25.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/32/741832/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/32/741832/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/8"}}},"commit":{"parents":[{"commit":"3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7","subject":"Merge \"Handle subnet without gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:50:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 23:02:19.000000000","tz":0},"subject":"[WIP] Migrate jobs to zuul v3 syntax","message":"[WIP] Migrate jobs to zuul v3 syntax\n\nChange-Id: I54d784fe0bded5f461ba9f26ed901ec042a6bd0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/01cfd8fc208aec252c11020304a57935b8df319c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/01cfd8fc208aec252c11020304a57935b8df319c"}]},"branch":"refs/heads/master"},"e4d8b43e086e0e271d4cec4b8ea90e106cb324fd":{"kind":"REWORK","_number":9,"created":"2020-07-20 01:27:27.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/32/741832/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/32/741832/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/9"}}},"commit":{"parents":[{"commit":"8eee0968b9b3a8b4c7bca2d0e1dbce465bb9bbdd","subject":"Use python3 by default in all jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8eee0968b9b3a8b4c7bca2d0e1dbce465bb9bbdd"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:50:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-20 01:27:21.000000000","tz":0},"subject":"[WIP] Migrate jobs to zuul v3 syntax","message":"[WIP] Migrate jobs to zuul v3 syntax\n\nChange-Id: I54d784fe0bded5f461ba9f26ed901ec042a6bd0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/e4d8b43e086e0e271d4cec4b8ea90e106cb324fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/e4d8b43e086e0e271d4cec4b8ea90e106cb324fd"}]},"branch":"refs/heads/master"},"3f766bb2b0688b173820eafea255a3fee27ea14d":{"kind":"REWORK","_number":10,"created":"2020-07-20 01:31:28.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/32/741832/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/32/741832/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/10"}}},"commit":{"parents":[{"commit":"8eee0968b9b3a8b4c7bca2d0e1dbce465bb9bbdd","subject":"Use python3 by default in all jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8eee0968b9b3a8b4c7bca2d0e1dbce465bb9bbdd"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:50:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-20 01:31:20.000000000","tz":0},"subject":"[WIP] Migrate jobs to zuul v3 syntax","message":"[WIP] Migrate jobs to zuul v3 syntax\n\nChange-Id: I54d784fe0bded5f461ba9f26ed901ec042a6bd0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3f766bb2b0688b173820eafea255a3fee27ea14d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3f766bb2b0688b173820eafea255a3fee27ea14d"}]},"branch":"refs/heads/master"},"08409761620e1122cb2b5b2627ac6beaff8b5ef9":{"kind":"REWORK","_number":11,"created":"2020-07-20 02:33:16.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/32/741832/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/32/741832/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/11"}}},"commit":{"parents":[{"commit":"8eee0968b9b3a8b4c7bca2d0e1dbce465bb9bbdd","subject":"Use python3 by default in all jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8eee0968b9b3a8b4c7bca2d0e1dbce465bb9bbdd"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:50:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-20 02:30:08.000000000","tz":0},"subject":"Migrate jobs to zuulv3 native","message":"Migrate jobs to zuulv3 native\n\nMigrate the following jobs to zuul v3:\n* kuryr-libnetwork-fullstack-py3\n* kuryr-libnetwork-kuryr_lib-src-fullstack\n* kuryr-libnetwork-install\n* kuryr-libnetwork-fullstack-ipv6-only\n\nRemove the following unused jobs:\n* kuryr-libnetwork-install-multinode\n\nChange-Id: I54d784fe0bded5f461ba9f26ed901ec042a6bd0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/08409761620e1122cb2b5b2627ac6beaff8b5ef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/08409761620e1122cb2b5b2627ac6beaff8b5ef9"}]},"branch":"refs/heads/master"},"055745136e5715f0f5bb1828cbcc46cf9be145ba":{"kind":"REWORK","_number":12,"created":"2020-07-22 02:06:54.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/32/741832/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/32/741832/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/32/741832/12"}}},"commit":{"parents":[{"commit":"8eee0968b9b3a8b4c7bca2d0e1dbce465bb9bbdd","subject":"Use python3 by default in all jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8eee0968b9b3a8b4c7bca2d0e1dbce465bb9bbdd"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 03:50:38.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-22 02:06:45.000000000","tz":0},"subject":"Migrate jobs to zuulv3 native","message":"Migrate jobs to zuulv3 native\n\nMigrate the following jobs to zuul v3:\n* kuryr-libnetwork-fullstack-py3\n* kuryr-libnetwork-kuryr_lib-src-fullstack\n* kuryr-libnetwork-install\n* kuryr-libnetwork-fullstack-ipv6-only\n\nRemove the following unused jobs:\n* kuryr-libnetwork-install-multinode\n\nChange-Id: I54d784fe0bded5f461ba9f26ed901ec042a6bd0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/055745136e5715f0f5bb1828cbcc46cf9be145ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/055745136e5715f0f5bb1828cbcc46cf9be145ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
