)]}'
{"id":"openstack%2Fkuryr-libnetwork~741856","triplet_id":"openstack%2Fkuryr-libnetwork~master~I823085b73ec816177a1f63c22eb4b7438bbd560a","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"I823085b73ec816177a1f63c22eb4b7438bbd560a","subject":"Use python3 by default in all jobs","status":"MERGED","created":"2020-07-19 22:33:20.000000000","updated":"2020-07-24 03:32:33.000000000","submitted":"2020-07-24 03:28:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741856-1595561336351-1d32c05a","meta_rev_id":"b59d3271d7a812a92ebd93bef6d059f8990a7b43","_number":741856,"virtual_id_number":741856,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 03:28:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-07-21 08:05:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-07-23 14:27:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-07-24 02:48:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-21 08:05:47.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-07-23 14:27:16.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2020-07-24 03:28:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a82d6cd99954bc9cde38f22f459c8d7a96d68616","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-19 22:33:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d931eed6480b85aa45155aed3866833cc95eb61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 23:20:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88e55158d8684777b0bd29e2f1b88766 : SUCCESS in 3m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/202bf4ff74b7410e938a5f5bf344f1a1 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35adffd1363b4b2581ae0ae80198dcb0 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b30f32b019d43dba5d48fa9cf26b9c8 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18243f6fe725428d868d59311c7eae89 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6418aa33b6934a59a8cf36fd460f5fa0 : SUCCESS in 3m 55s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/f0fa2dcf135c4c06a6de39111ae837e9 : SUCCESS in 31m 41s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/34df50d0777144e69fd446042bfe3117 : FAILURE in 40m 15s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/2ad6cd025deb48759074953533e8138d : FAILURE in 34m 13s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/d8e4323260d243f2bd6c5f186363cd0f : SUCCESS in 35m 01s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/d14ed15d2a8441638f8e9814bf4e6f12 : FAILURE in 33m 27s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/09a0ee4e9c0249099ac736ead0b017ea : SUCCESS in 33m 47s","accounts_in_message":[],"_revision_number":1},{"id":"72b99c926d05c49ac548e016389e1754fa813cea","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-20 01:00:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eebf08b9300445d052e7891de40abd5eadc1fbb2","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-20 01:27:27.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a92dd746dfa9a0b8692c2ec05b64ab3a9a5f3ea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 01:57:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/481cb35f5a30485880664fdc8e98ed68 : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3640c2a6127439086d7b8becb7ef338 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2211d5805404fc6a3f970626180edc1 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a72f48c6b18941299c38d6039142b077 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ec1d21b5c7b4c68b39780525c6e847f : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab95ae97b4c4c40baf427554258cb6d : SUCCESS in 4m 11s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/8cf86b6e375d4406b2366576f4f5a609 : SUCCESS in 19m 52s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/0caaa9f2992f4d9ab1e1e850c6356631 : SUCCESS in 27m 43s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/77e76e45a3304c65bcdd913fd1d6dbb7 : FAILURE in 27m 26s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/b4874171bf964e98bbd08fd96cc8d954 : SUCCESS in 21m 20s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/8775f3acf6c947da887200a6c05f5c9a : FAILURE in 21m 53s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/111879e44a0d49a38b0ed7bf957e2fb2 : SUCCESS in 27m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2570ac972ff5cf5dc738583f534e65f4b547c4d9","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-21 08:05:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"85a503a4b21367f9b4e138b74f990df35b52ef1f","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2020-07-23 14:27:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e567013de9c88c9819f5bd339a8734058666f64c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-24 02:48:33.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ba9a030d54b562ee152f7bdc54c65b4a2a6832be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 02:48:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"65d99737cfd29d0f8cfa3bb023e10254e9a1b386","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 03:28:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8b1ad0d58267e17b97a77dbadc8deef77b214181","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 03:28:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7549357794942bf847f1ff904ccbba2 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a21fef071df43d9bac7216326be4e0a : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1203bfacc7f34e4a827ea9b0be1a22e4 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adb15a0ccff8431492b959d03dd4afc8 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7933997e88047178312343429b941d6 : SUCCESS in 4m 49s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/36be5e73bd414e7a93afcd1640c32051 : SUCCESS in 29m 57s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/d0f4dcee2dab47aaade568108e8e0c58 : SUCCESS in 34m 02s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/3f6498ff6ac04d4e9119a8832893b21a : SUCCESS in 34m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b59d3271d7a812a92ebd93bef6d059f8990a7b43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 03:32:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d780a16c7cc4b9b8690e66cba4d5c3b : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8eee0968b9b3a8b4c7bca2d0e1dbce465bb9bbdd","revisions":{"6bf545ff5c8640d8259930f5d9104634c9b73e1a":{"kind":"REWORK","_number":1,"created":"2020-07-19 22:33:20.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/56/741856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/56/741856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/56/741856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/56/741856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/56/741856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/56/741856/1"}}},"commit":{"parents":[{"commit":"3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7","subject":"Merge \"Handle subnet without gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 22:33:03.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 22:33:03.000000000","tz":0},"subject":"Use python3 by default in all jobs","message":"Use python3 by default in all jobs\n\nChange-Id: I823085b73ec816177a1f63c22eb4b7438bbd560a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/6bf545ff5c8640d8259930f5d9104634c9b73e1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/6bf545ff5c8640d8259930f5d9104634c9b73e1a"}]},"branch":"refs/heads/master"},"8eee0968b9b3a8b4c7bca2d0e1dbce465bb9bbdd":{"kind":"NO_CHANGE","_number":2,"created":"2020-07-20 01:27:27.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/56/741856/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/56/741856/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/56/741856/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/56/741856/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/56/741856/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/56/741856/2"}}},"commit":{"parents":[{"commit":"3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7","subject":"Merge \"Handle subnet without gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3c1a92403fbe64ee138bf9c676c39aedf1e2e5c7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-19 22:33:03.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-20 01:12:50.000000000","tz":0},"subject":"Use python3 by default in all jobs","message":"Use python3 by default in all jobs\n\nChange-Id: I823085b73ec816177a1f63c22eb4b7438bbd560a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8eee0968b9b3a8b4c7bca2d0e1dbce465bb9bbdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8eee0968b9b3a8b4c7bca2d0e1dbce465bb9bbdd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
