)]}'
{"id":"openstack%2Fkuryr-libnetwork~743047","triplet_id":"openstack%2Fkuryr-libnetwork~master~I8fa392718c2a747a99639d0bd1fa1e2e97a7b47a","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"I8fa392718c2a747a99639d0bd1fa1e2e97a7b47a","subject":"Remove unnecessary \"run\" and \"post-run\"","status":"MERGED","created":"2020-07-25 23:06:26.000000000","updated":"2020-07-27 08:38:40.000000000","submitted":"2020-07-27 08:34:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743047-1595838865502-bdae804c","meta_rev_id":"6d9c4b2598c2bccee6da9d1a70f5126f7d8fea2f","_number":743047,"virtual_id_number":743047,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-27 08:34:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-07-27 07:49:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2020-07-27 07:49:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"date":"2020-07-26 00:41:47.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-26 00:41:47.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-07-27 07:49:41.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-07-27 08:34:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7199a30501c73c1c6541556fe62ecd377887b1b2","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-07-25 23:06:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"508db12d8d14e9de3fa05193c54d7898bc88266a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 23:39:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08e157d5db5c4c00b4f1ded985fd8254 : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d4d7b8d602d4fd187de31330381faec : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9aec217ed6944df9a07c1696b545536 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81d3b810c33f4848acd3f26ae64633a8 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d3c5eff2d614dc08c539cf36a34978b : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7a44992b5a74cabbd69006bb91e4fae : SUCCESS in 4m 49s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/ca64c18a1cd54de28d7261cdcfa77a6c : SUCCESS in 26m 45s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/302a4d9279c644de800ee3eeafc20a77 : SUCCESS in 30m 11s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/915eb3376c054fcd80009eb5add4a832 : FAILURE in 32m 00s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/ed3e145ea6fa4ae58a3490a431cf9ea9 : SUCCESS in 31m 59s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/788274970df24146a0a484b444a70fdc : FAILURE in 30m 39s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/deb2b391a0d8431197228d49a16f0222 : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":1},{"id":"381720585bd3b7644556379020df1ac138c3cd29","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-27 07:49:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9d318a30f19d39f9d9ac13c144dc0b5f51cea1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 07:49:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f23ec874fd28322e617a559674ff1502416aedf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 08:34:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d141835e92b44f38594a94bcb4f5599 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1bff86dd7e446859d3a5534ebccdc89 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9880a90fd71644af8b7b0a70448745e2 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48ede2f7e0034016ab2c7bf684e4cf79 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c0d13522d3f4fea84ab65a94d0e243d : SUCCESS in 5m 31s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/6e263cd9547e4a02b99c30d42c4a52bc : SUCCESS in 38m 10s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/ec1527c888b1421b8bb9e91723a169a4 : SUCCESS in 28m 08s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/590b6c17e7734faf90a91057fdfb7bee : SUCCESS in 33m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f72f98dc2c922b7e2d0c9f22a00d2163107bb079","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 08:34:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6d9c4b2598c2bccee6da9d1a70f5126f7d8fea2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 08:38:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb83835f5d14b09965a79c16a4188f2 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b096b19a5fcb2633160bbd9f03a3b69acf379057","revisions":{"b096b19a5fcb2633160bbd9f03a3b69acf379057":{"kind":"REWORK","_number":1,"created":"2020-07-25 23:06:26.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/47/743047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/47/743047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/47/743047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/47/743047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/47/743047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/47/743047/1"}}},"commit":{"parents":[{"commit":"055745136e5715f0f5bb1828cbcc46cf9be145ba","subject":"Migrate jobs to zuulv3 native","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/055745136e5715f0f5bb1828cbcc46cf9be145ba"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-25 23:04:13.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-07-25 23:06:26.000000000","tz":0},"subject":"Remove unnecessary \"run\" and \"post-run\"","message":"Remove unnecessary \"run\" and \"post-run\"\n\nChange-Id: I8fa392718c2a747a99639d0bd1fa1e2e97a7b47a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/b096b19a5fcb2633160bbd9f03a3b69acf379057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/b096b19a5fcb2633160bbd9f03a3b69acf379057"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
