)]}'
{"id":"openstack%2Fkuryr-libnetwork~755914","triplet_id":"openstack%2Fkuryr-libnetwork~master~Ia635ecb0eaea3a400a0b03031efec354d0c3c77e","project":"openstack/kuryr-libnetwork","branch":"master","topic":"optimization/pip-no-cache-dir","hashtags":[],"change_id":"Ia635ecb0eaea3a400a0b03031efec354d0c3c77e","subject":"optimize size and time using --no-cache-dir","status":"MERGED","created":"2020-10-03 06:22:23.000000000","updated":"2020-10-09 17:34:51.000000000","submitted":"2020-10-09 17:32:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755914-1602264777614-0a98ea85","meta_rev_id":"06ff9b26e4dcd73e09381ea67f6682aecd1f9b1f","_number":755914,"virtual_id_number":755914,"owner":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-09 17:32:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2020-10-09 13:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2020-10-09 16:44:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-10-09 16:44:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 13:05:04.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-10-09 16:44:43.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-10-09 17:32:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c333501d466a5bb965b6d7d601f5ce534debeabb","author":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"date":"2020-10-03 06:22:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06fd7ff5c480da68ee89278e42889628695868d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 07:09:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06543b4f7cbf40cda64440fa7537b348 : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/880135566c4e4fdba63282320b806bbb : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1975bf0e1ea40408e1323375e0297ab : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aca67d4a05664becab0e906d6995ec07 : FAILURE in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01f7b60438eb49dfa1e5bbe997396077 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47a071dfc60748fa8d7c0353fd534ec5 : SUCCESS in 4m 08s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/2294d5bdc99f4bf9ad7c9aac22bd7ed3 : SUCCESS in 35m 54s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/2149bc00dc6c48eb8109b09ce7447964 : FAILURE in 36m 35s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/68017f8542eb4ecb9c7e8693ac70f8d6 : FAILURE in 41m 27s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/bc65456de2574008a2a88d7396c420f2 : FAILURE in 35m 28s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/7cc20870eade4a30a6ddb8f07ea63a23 : FAILURE in 37m 24s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/e7c3a254c13946c4b498a71e1ca7f8b4 : FAILURE in 31m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b16cc62b3ca149dec0f240ae3e5764905b7181cd","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-10-09 13:05:04.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\n\nThis dockerfile is suboptimal anyway, but this won\u0027t hurt.","accounts_in_message":[],"_revision_number":1},{"id":"12f6269dc3239ef022a33cb4b7b5b94ba5a7d209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 14:18:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0b4ab89bdc4499c86d4e7318c2dd4f8 : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0593aa2716874ccfacf44522bee1a9b6 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a24d490519456fb1ac9b6adb9729d2 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4532f6eaa7c64585aef89722faa89456 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b427203b18e948a3bc25c8b882557b37 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eca187bdaf534802869738354c5cb968 : SUCCESS in 4m 32s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/669e33774ab040d999ddf1d37cbe92f9 : SUCCESS in 33m 44s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/fd2fbd21bc084a40a6c6dc557ff16a72 : SUCCESS in 37m 53s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/c544cee92d21486d938d53d283de4988 : FAILURE in 36m 51s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/dc9441172aa4405990058e804143d91e : SUCCESS in 36m 31s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/58198b493e5e4769ab9fa0b7b88238fa : FAILURE in 31m 05s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/ea3f028ea04f4ec199493348bbf42cbd : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ba272e67c2a7bef533a96fe40cf14f8954192a8b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-10-09 16:44:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0bbbcb65d9df46f63b45e52424beadb2059ca584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 16:44:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7de106a989dfff2dc880e7bf1c3c103ae5fb00a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 17:32:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf4e38ba65b54db488466a0e9f7e0c50 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8618d0c17f7d4daea344913200a13c33 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d2ce306f6ec4950aacba830438f84b8 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d5bbaca44f34f028c2f958c02c02717 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e302ba5804584cfbb88e2250d120f029 : SUCCESS in 4m 00s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/9cde1bea2969494ba7a109310ae8bb10 : SUCCESS in 38m 54s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/874e594902cd49cdb6cb9bc10ffe42cf : SUCCESS in 34m 26s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/aca338ea34ee4ada8e15eb8a7582afe6 : SUCCESS in 40m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f867d108e8dacac03a0564d7e04205d6c881d436","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 17:32:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f8b19f795ace20a5481843c53c91c7ead7bb7a47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 17:34:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b40caa6fe8d4f7aad5ed7bab62103d3 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"952e881922943f72bc0e634571146cea3b54a819","revisions":{"952e881922943f72bc0e634571146cea3b54a819":{"kind":"REWORK","_number":1,"created":"2020-10-03 06:22:23.000000000","uploader":{"_account_id":31020,"name":"Pratik raj","email":"Rajpratik71@gmail.com","username":"rajpratik71"},"ref":"refs/changes/14/755914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/14/755914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/14/755914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/14/755914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/14/755914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/14/755914/1"}}},"commit":{"parents":[{"commit":"c4d74c2d89a2e6ee11c2241d7eab64dcc2f55c6a","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/c4d74c2d89a2e6ee11c2241d7eab64dcc2f55c6a"}]}],"author":{"name":"Pratik Raj","email":"rajpratik71@gmail.com","date":"2020-10-03 06:21:54.000000000","tz":330},"committer":{"name":"Pratik Raj","email":"rajpratik71@gmail.com","date":"2020-10-03 06:22:16.000000000","tz":330},"subject":"optimize size and time using --no-cache-dir","message":"optimize size and time using --no-cache-dir\n\nUsing --no-cache-dir flag in pip install ,make sure dowloaded packages\nby pip don\u0027t cached on system . This is a best practise which make sure\nto fetch ftom repo instead of using local cached one . Further , in case\nof Docker Containers , by restricing caching , we can reduce image size.\nIn term of stats , it depends upon the number of python packages\nmultiplied by their respective size . e.g for heavy packages with a lot\nof dependencies it reduce a lot by don\u0027t caching pip packages.\n\nFurther , more detail information can be found at\n\nhttps://medium.com/sciforce/strategies-of-docker-images-optimization-2ca9cc5719b6\n\nChange-Id: Ia635ecb0eaea3a400a0b03031efec354d0c3c77e\nSigned-off-by: Pratik Raj \u003crajpratik71@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/952e881922943f72bc0e634571146cea3b54a819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/952e881922943f72bc0e634571146cea3b54a819"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
