)]}'
{"id":"openstack%2Fkuryr-libnetwork~769442","triplet_id":"openstack%2Fkuryr-libnetwork~master~Iac9a3421c23370f19f57d6aa84c2862ccfce2a94","project":"openstack/kuryr-libnetwork","branch":"master","topic":"fix-relmgt-pip-doc","hashtags":[],"change_id":"Iac9a3421c23370f19f57d6aa84c2862ccfce2a94","subject":"Add doc/requirements","status":"MERGED","created":"2021-01-06 01:27:02.000000000","updated":"2021-01-11 17:33:19.000000000","submitted":"2021-01-11 17:32:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769442-fix-relmgt-pip-doc","meta_rev_id":"09e414ae7f30b01f7bb53691b7c761173825f7a2","_number":769442,"virtual_id_number":769442,"owner":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-11 17:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2021-01-07 06:28:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2021-01-11 16:28:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-07 07:38:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2021-01-11 16:28:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-06 02:00:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-06 02:50:33.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2021-01-06 02:50:33.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2021-01-06 02:50:33.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2021-01-07 06:28:03.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"00a489ed5dcdf01309a3db27da15d78a346a9321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-01-06 01:27:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3558a91c34fa5fc3525677b4a359ebd85b2fb7b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 02:00:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17092cc514184dcdaad095c2b32c3a64 : SUCCESS in 3m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7afb033833864c1c8fe505c9074a728c : FAILURE in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3efbcb94a1e4de487266535e8d177fc : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/642d1f08d51e4cd0ab221374819259c5 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a79e59ab0ddb4372be8fecb14bbf54a8 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/071f33ebf55f4724ab21764287373ebb : SUCCESS in 3m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a579a5f7f434231816ef7442be9b94f : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6cac382fa7d42c4b5a38170a9a8108b : SUCCESS in 3m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b7d9e517745b4e30b0ab2b95eb903d6c : SUCCESS in 1m 55s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/710b3ff7d492408dafd22c4376ad282d : SUCCESS in 27m 14s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/e53080581ff447b296bfa6349f031ddd : FAILURE in 29m 22s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/f1b7412f1ca045159f46ac8114e643e3 : FAILURE in 30m 01s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/1db54f0af143495aaebcbfcd471e2724 : SUCCESS in 29m 39s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/3c858f48cb4b4335ab6d84b7dee78448 : FAILURE in 32m 09s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/94cb9342c04f4502a7caa7568cf97c51 : SUCCESS in 27m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d96866e2b4c881d59e209300ea4fe084a0b5afca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-01-06 02:05:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e4ed6151ad65544adee4748167631bbc142ae176","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-01-06 02:06:02.000000000","message":"Topic fix-lc removed","accounts_in_message":[],"_revision_number":2},{"id":"1cb5ca650143eed072040857cbc20698b09323b8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-01-06 02:07:03.000000000","message":"Topic set to fix-relmgt-pip-doc","accounts_in_message":[],"_revision_number":2},{"id":"623f25f1f200a27a27650335428ae1bc9fe77ce8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 02:49:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0a8f14a9a7849babe7362752bcb1dfb : SUCCESS in 3m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5656fea30f64625b96668416072151a : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97e571272ba740ed9cdcee983e1f41c2 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/261dba55fd6d4920b2912ca57858ece8 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fa5b5cff4d643f0ae04a2801e0b689e : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9fcdd1c75f040ea80e73a45c7cf6a16 : SUCCESS in 3m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9bf1664e8104d6da61fa268ae5f8780 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44ad13e31c2d470fa24a39493d0581ef : SUCCESS in 3m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90b8a91496004ee8bc42d675c6ac0e35 : SUCCESS in 1m 44s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/08af8891c28d4ca28f61516bfa3554b5 : SUCCESS in 24m 36s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/0fe3d2f1d1dd4dd78f960502f5f615e5 : SUCCESS in 37m 01s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/8494a1294bda468287682fa735dd53d1 : FAILURE in 29m 35s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/1f4751c322ee42a2bacab06a3e57035a : SUCCESS in 34m 22s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/abcccb537af04c2c872384812a06ab95 : FAILURE in 38m 09s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/3bd2063e760d4f80a516ff392c628f78 : SUCCESS in 31m 41s","accounts_in_message":[],"_revision_number":2},{"id":"3f24bf2b2cff1673136bea904fadcb0d896d7767","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-01-07 06:28:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"19c1e85a8e4c93ff532ac3f599634092419d4a60","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-01-07 07:38:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b6f3417c09925c02bb9bb2458b2a4ab6558db082","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-01-11 16:28:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4e2ce0d384dd0b2a4a5816c90b0b2b11908eb6a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:29:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"950afbfe843243cb4fb0169d70c15ba89e6673f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 17:32:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58878c58bf134de499040d0a3d5fc6a3 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec737f6c46024462b0a0dbf1c7f9cec2 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b0b4f8583f24cfbbc92a57e5e054d46 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f22ad70565847cc9216043963d3b5f4 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eea72c3fe38b4b1b8d248e0387fe1cb7 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0accb03024048ed9456defd486aaedc : SUCCESS in 5m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/65d30b4e59bc48258a02f5e00407764c : SUCCESS in 2m 47s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/3797d06eac86484b8970c01c80a833b9 : SUCCESS in 59m 46s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/89519d0a7a514a0aaae15fa97ac953ee : SUCCESS in 39m 50s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/0882eb76fb414ad3afc6c4862c8129ee : SUCCESS in 33m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3ba2d22565b9e106c8b21c81c8205e1238e82e66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 17:32:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"09e414ae7f30b01f7bb53691b7c761173825f7a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 17:33:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b81e942c985248199bd4b0df367ca0e8 : SUCCESS in 49s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8aa7952215e4cc2953de4afc54a4743 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8f7866809240acd52667f9e47a5df4a700d9290b","revisions":{"e6521d0ff30ff2b6f04a69145743c2518a55b738":{"kind":"REWORK","_number":1,"created":"2021-01-06 01:27:02.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/42/769442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/42/769442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/42/769442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/42/769442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/42/769442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/42/769442/1"}}},"commit":{"parents":[{"commit":"d68264765d663299a737f9aa63202e6da88c2355","subject":"Fix lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/d68264765d663299a737f9aa63202e6da88c2355"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-01-06 01:25:15.000000000","tz":-180},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-01-06 01:25:15.000000000","tz":-180},"subject":"Add doc/requirements","message":"Add doc/requirements\n\nSpecify doc requirements needs to be moved out from\ntest-requirements.txt to in doc/requirements.txt\nto avoid problems with the pip resolver as announced by\nthe release team [1].\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019611.html\n\nChange-Id: Iac9a3421c23370f19f57d6aa84c2862ccfce2a94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/e6521d0ff30ff2b6f04a69145743c2518a55b738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/e6521d0ff30ff2b6f04a69145743c2518a55b738"}]},"branch":"refs/heads/master"},"8f7866809240acd52667f9e47a5df4a700d9290b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-06 02:05:31.000000000","uploader":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"ref":"refs/changes/42/769442/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/42/769442/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/42/769442/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/42/769442/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/42/769442/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/42/769442/2"}}},"commit":{"parents":[{"commit":"e7bc12e5cbf9c9abe2ef69ffa3113558d60c88a6","subject":"Fix lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/e7bc12e5cbf9c9abe2ef69ffa3113558d60c88a6"}]}],"author":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-01-06 01:25:15.000000000","tz":-180},"committer":{"name":"Maysa Macedo","email":"maysa.macedo95@gmail.com","date":"2021-01-06 02:03:40.000000000","tz":-180},"subject":"Add doc/requirements","message":"Add doc/requirements\n\nSpecify doc requirements needs to be moved out from\ntest-requirements.txt to in doc/requirements.txt\nto avoid problems with the pip resolver as announced by\nthe release team [1].\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019611.html\n\nChange-Id: Iac9a3421c23370f19f57d6aa84c2862ccfce2a94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8f7866809240acd52667f9e47a5df4a700d9290b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8f7866809240acd52667f9e47a5df4a700d9290b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
