)]}'
{"id":"openstack%2Fkuryr-libnetwork~789076","triplet_id":"openstack%2Fkuryr-libnetwork~master~Ief784b30ea4786b283f38b85950b7788592f67e8","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"Ief784b30ea4786b283f38b85950b7788592f67e8","subject":"Dropping lower constraints testing","status":"NEW","created":"2021-05-02 02:29:03.000000000","updated":"2021-05-11 10:35:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e016fa79d795aaa67475a5da3ddca7fd38968301","_number":789076,"virtual_id_number":789076,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-02 03:02:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":-1,"date":"2021-05-07 11:49:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":-1,"date":"2021-05-11 10:35:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-02 03:02:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-07 11:49:23.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2021-05-11 10:35:12.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"}],"messages":[{"id":"ba415b683ee580a63c0f418b1e4aaa991cb14f6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-05-02 02:29:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9c752c7099cab3a4263e635e254e2248cb34e2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-02 03:02:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/599529f8312e4b6eb2106c14718db5e3 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaeccc6707634103ad487a654154a0fb : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc1e0c1baa4545848271900ce1d4f6ad : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/487e2d0964c94281a10640c40134adde : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d125b7a7e4d14a6e8ccab0e883e1726c : SUCCESS in 4m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/038a9b6e69484ebca4534995658f3f65 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884ea71d0d7341278b987486c7fe8a8d : SUCCESS in 4m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d8572fbbef9439bbe468cc71b37fb23 : SUCCESS in 3m 16s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/41df82e44c1648febdde5e96afa66f53 : SUCCESS in 31m 29s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/aacdabb4ad3742beb5fc5999453a3ebb : SUCCESS in 26m 42s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/4a0680920eda40cab6096a6e6f3c9b5a : FAILURE in 21m 37s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/977af9f0d21842cdbe90e8513d842e74 : SUCCESS in 28m 10s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/140fc4825df6477c9ba0584bc38aa45a : FAILURE in 28m 14s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/620baa70c9094bcf86a35266be03fd28 : SUCCESS in 31m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ceee546820bc4877114d231952c64818cddc921f","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2021-05-07 11:49:23.000000000","message":"Patch Set 1: Code-Review-1\n\nLooking at results of Ie4e3366ad01e219971595464122a6b262e55cb91, the lower-constraints job is fine here as we put some work to get it fixed. What\u0027s the point of this patch?","accounts_in_message":[],"_revision_number":1},{"id":"e016fa79d795aaa67475a5da3ddca7fd38968301","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2021-05-11 10:35:12.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Looking at results of Ie4e3366ad01e219971595464122a6b262e55cb91, the lower-constraints job is fine here as we put some work to get it fixed. What\u0027s the point of this patch?\n\n+1. It might be interesting to share the errors you got, so we can better evaluate the need for dropping it.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"53087dde7a221137987a4422feb3a6d6c47023ae","revisions":{"53087dde7a221137987a4422feb3a6d6c47023ae":{"kind":"REWORK","_number":1,"created":"2021-05-02 02:29:03.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/76/789076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/76/789076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/76/789076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/76/789076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/76/789076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/76/789076/1"}}},"commit":{"parents":[{"commit":"f2edb0c54a1b2303bd1c7a45b3e7342daa854b99","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/f2edb0c54a1b2303bd1c7a45b3e7342daa854b99"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-05-02 02:28:25.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-05-02 02:28:58.000000000","tz":480},"subject":"Dropping lower constraints testing","message":"Dropping lower constraints testing\n\nWe facing errors related to the new pip resolver, this\ntopic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nI propose to drop this test because the complexity and recurring pain needed\nto maintain that now exceeds the benefits provided by this mechanismes.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nChange-Id: Ief784b30ea4786b283f38b85950b7788592f67e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/53087dde7a221137987a4422feb3a6d6c47023ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/53087dde7a221137987a4422feb3a6d6c47023ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
