)]}'
{"id":"openstack%2Fkuryr-libnetwork~850983","triplet_id":"openstack%2Fkuryr-libnetwork~master~I0be74cd810fce7ac98d97a4d362fc92190721bb5","project":"openstack/kuryr-libnetwork","branch":"master","attention_set":{},"removed_from_attention_set":{"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2022-08-12 06:09:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0be74cd810fce7ac98d97a4d362fc92190721bb5","subject":"Use TOX_CONSTRAINTS_FILE","status":"MERGED","created":"2022-07-26 09:15:56.000000000","updated":"2022-08-12 06:10:10.000000000","submitted":"2022-08-12 06:09:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850983","meta_rev_id":"6e03620c4ce85b5c900ba32d7a98cf9411720f72","_number":850983,"virtual_id_number":850983,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-12 06:09:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"recommended":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2022-07-27 08:26:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2022-08-12 05:39:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2022-07-26 09:34:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2022-08-12 05:39:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-26 09:34:07.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2022-07-26 09:54:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-27 08:26:53.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2022-08-12 05:39:39.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"807681ff276e8e2caf1648e08d812b6cde47548c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-07-26 09:15:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b20c71001cf8b2b205a8dac0d0e52c59c1699df","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-07-26 09:34:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"174d6477aa657375f5ba39b9e10c15d4bc788794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 09:54:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e4a56579ffa41f09d2910326aafdf0e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b52afd97bb8a4b4b9c9597751a2502e9 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f97fe884ea4183a8b983d8c2e9b6e0 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37491c85d4a84dc69bfac03125163f6f : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28b61e83007e4d858f20d1c9a09cd2d6 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9a8d93e2b90405796a87a1ffe64546d : SUCCESS in 4m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a428c89cef9f4d77b6586e66b63ec71a : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/281b59521ba643d6a21f4844cc3ea158 : SUCCESS in 4m 21s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/4aed7f3d5db842d6b689fede7a2597a8 : SUCCESS in 33m 48s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/1bdd3adabbbd4da0beaecc96734b673a : SUCCESS in 33m 37s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/bf726a8db5274cae93df10173d98ecc8 : FAILURE in 30m 16s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/19927b1f1ea443d49597ca22107b3d98 : SUCCESS in 36m 52s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/263b3871c3ce4a8cbc92dc669cc7da47 : FAILURE in 29m 11s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/7e25349f9f6f4b6dad05fa0c78c4d63c : SUCCESS in 35m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ad1fb754dea1260640a9996acbe914bff2cf7fb5","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-07-27 08:26:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"451e23061a5c799c48629160049801c5833d8dff","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2022-08-12 05:39:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3256d69327b7ba04f0ec124d5fbf52c2b684e919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 05:40:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d709be7f72dbe066c71f03089b6c448b218c4c7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 06:09:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10ec6277f0b84096910e942b3dbb549b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93dbb1c5eeb544f9b4dadd4f0a3e8c86 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdc7b7d2d0274bde855c7d386e61d9b0 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bf72309c22146deab9bd5507da35371 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f73f180b8be546f6bf12161e04efc4f7 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15f48f527d154bed9171971a81d96bbb : SUCCESS in 3m 44s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/9e057566fd8444a98e397172d6d228e3 : SUCCESS in 28m 02s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/ec5b95bbf4934a1087fc28ab1b1b7d85 : SUCCESS in 27m 43s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/4531d7dc5e0c4f17af079b8543dc7168 : SUCCESS in 26m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3c91eb0e0f902da290ef299751fd7f7de4696aa2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 06:09:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6e03620c4ce85b5c900ba32d7a98cf9411720f72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 06:10:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f57e67c525c43498c057c5e43f7a753\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bff630dd805046adba4e817535343353 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f82636cf3214730843325e8d787a50f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4daa56797dd18870e61977cf1248abb117ca009c","revisions":{"4daa56797dd18870e61977cf1248abb117ca009c":{"kind":"REWORK","_number":1,"created":"2022-07-26 09:15:56.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/83/850983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/83/850983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/83/850983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/83/850983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/83/850983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/83/850983/1"}}},"commit":{"parents":[{"commit":"f73cad32770a6a00c9ee55aee6fbde2b0f87e25d","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/f73cad32770a6a00c9ee55aee6fbde2b0f87e25d"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-26 09:14:56.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-26 09:15:05.000000000","tz":480},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I0be74cd810fce7ac98d97a4d362fc92190721bb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/4daa56797dd18870e61977cf1248abb117ca009c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/4daa56797dd18870e61977cf1248abb117ca009c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[]}
