)]}'
{"id":"openstack%2Fkuryr-libnetwork~852310","triplet_id":"openstack%2Fkuryr-libnetwork~master~Ia17a0b0903f0d0257aeed2fb220b595d26d3780c","project":"openstack/kuryr-libnetwork","branch":"master","hashtags":[],"change_id":"Ia17a0b0903f0d0257aeed2fb220b595d26d3780c","subject":"remove unicode from code","status":"ABANDONED","created":"2022-08-06 00:54:53.000000000","updated":"2022-08-19 05:50:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8142295899ab42449145a53222e9d080c32e678e","_number":852310,"virtual_id_number":852310,"owner":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-06 01:02:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-06 01:02:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36907011095be3a27075fb8d326d669f9ce26072","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-06 00:54:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd5f100c5f72f3727a7272eb2e22f081f2fbcc88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 01:02:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042b73700db4423186747f6c7d188bae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17ab39a229654135a746599132788350 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36fba289a5464203822cd10dd1aa430e : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8142295899ab42449145a53222e9d080c32e678e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-19 05:50:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c8e447d7aff842c22afac58e93e84c66bc28b45","revisions":{"6c8e447d7aff842c22afac58e93e84c66bc28b45":{"kind":"REWORK","_number":1,"created":"2022-08-06 00:54:53.000000000","uploader":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"ref":"refs/changes/10/852310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/10/852310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/10/852310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/10/852310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/10/852310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/10/852310/1"}}},"commit":{"parents":[{"commit":"602c545268683b7cf5ebded0dbbdcad2bd4994fa","subject":"Merge \"Revert \"Fix the bindir path\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/602c545268683b7cf5ebded0dbbdcad2bd4994fa"}]}],"author":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-06 00:54:29.000000000","tz":480},"committer":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-06 00:54:37.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: Ia17a0b0903f0d0257aeed2fb220b595d26d3780c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/6c8e447d7aff842c22afac58e93e84c66bc28b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/6c8e447d7aff842c22afac58e93e84c66bc28b45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
