)]}'
{"id":"openstack%2Fkuryr-libnetwork~852325","triplet_id":"openstack%2Fkuryr-libnetwork~master~I179e635563ca79902a52d442ef44b9c7136c5d4c","project":"openstack/kuryr-libnetwork","branch":"master","attention_set":{},"removed_from_attention_set":{"35067":{"account":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"last_update":"2023-08-08 09:39:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I179e635563ca79902a52d442ef44b9c7136c5d4c","subject":"Use py3 as the default runtime for tox","status":"ABANDONED","created":"2022-08-06 01:36:11.000000000","updated":"2023-08-08 09:39:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fc8ed6fbedad3965a5ade84f1fee8d582f3cbe06","_number":852325,"virtual_id_number":852325,"owner":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-06 02:22:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-06 02:22:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a2036dab266e70cdb756a688e7b1abc1c22e700","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-06 01:36:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60121f80175abc1482e22a9ad9c18165994e302b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 02:22:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d90608483870482fa599166849f5d69b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7f0f77c4a254569b62ca36b1a577f75 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1969267aece48c986741f3c5aea1bc3 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea70fea23cf8406f800bf127435aa020 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d777892600ad4aaa9e0ff93e725be230 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/580baa67abcb4ccd9c92a61d91427d6b : SUCCESS in 4m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2a705469230445093fbe0d372f34c32 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fbd24e1e70447d8e61a7c605a16e7a : SUCCESS in 4m 21s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/c1a7ebad0aa045458575a7cad592e207 : SUCCESS in 29m 51s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/74a0fa3e681640ba9928d8fef859dd16 : SUCCESS in 40m 46s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/0e937f21f79a4e15afe08eccddd1e457 : FAILURE in 29m 07s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/a5a99c0972804b6a8016cee836f73b1b : FAILURE in 27m 26s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/e59e18fe17e64712a19f651cf267c457 : FAILURE in 28m 11s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/5f8262c7781d4b38a967b206046e1616 : SUCCESS in 26m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fc8ed6fbedad3965a5ade84f1fee8d582f3cbe06","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2023-08-08 09:39:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3723fb799a0bd37e2fb7a7b8d314c8465abe3b7","revisions":{"c3723fb799a0bd37e2fb7a7b8d314c8465abe3b7":{"kind":"REWORK","_number":1,"created":"2022-08-06 01:36:11.000000000","uploader":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"ref":"refs/changes/25/852325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/25/852325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/25/852325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/25/852325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/25/852325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/25/852325/1"}}},"commit":{"parents":[{"commit":"6c8e447d7aff842c22afac58e93e84c66bc28b45","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/6c8e447d7aff842c22afac58e93e84c66bc28b45"}]}],"author":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-06 01:35:22.000000000","tz":480},"committer":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-06 01:35:29.000000000","tz":480},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: I179e635563ca79902a52d442ef44b9c7136c5d4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/c3723fb799a0bd37e2fb7a7b8d314c8465abe3b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/c3723fb799a0bd37e2fb7a7b8d314c8465abe3b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
