)]}'
{"id":"openstack%2Fkuryr-libnetwork~852333","triplet_id":"openstack%2Fkuryr-libnetwork~master~Ibc07a8f3ecde14377a1afe33adb4d41307a00b2c","project":"openstack/kuryr-libnetwork","branch":"master","attention_set":{"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2024-02-26 10:27:49.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ibc07a8f3ecde14377a1afe33adb4d41307a00b2c","subject":"Use py3 as the default runtime for tox","status":"NEW","created":"2022-08-06 09:06:23.000000000","updated":"2024-02-26 10:27:49.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"893adba3eb2585ca0cfd80aad7d3a21a8a581d4b","_number":852333,"virtual_id_number":852333,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-26 10:27:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-06 09:46:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1aa0888bf7d3b00dd633d7e00787fc7c540a3b82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-06 09:06:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a70cf6d1bcfd35b8c5a77a32f0e61435ec1087a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 09:46:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b17ae50e1896497bafadd929a1e52379\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa10fbd41df84fb897cb96d6db998d33 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59ac022452794c86ac99dca199eed99b : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed314c0450e246e7825b577eb1e17a1f : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/614ef0a2eb3e414da2f6b9dd6862761a : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2126acbff30b437ea6c1024a128eddd7 : SUCCESS in 4m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2ed771da23044b1822737f9e6f89183 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1ac77540fa48649d7998ba59281882 : SUCCESS in 3m 35s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/e53edb99f53f4eaf9a4f290841047f32 : SUCCESS in 20m 46s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/b89f972897424e03809560a39409aa15 : SUCCESS in 38m 55s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/bef20ea2c4c3446fb34a267791ecaff8 : FAILURE in 25m 19s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/7ec1cb648dc648848f9736a51786ab1f : FAILURE in 23m 42s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/69d8e10847d44dbeba72f1cc89bbf38a : FAILURE in 27m 01s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/dd7624a4696f4b68b8286a0aef7fc5e3 : SUCCESS in 23m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e7700ae4ce7da842e44f4e3dae8e94786a81100c","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-09 09:46:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6faa25f6b83d5934a28ff3e9433fd9c2aeaa0fbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 10:34:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b0c84207b5547799d2477a5c5b6ee13\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/713509d834de484ea39f79708f75c456 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41faaac65d6641188822f83481dace32 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7676de47f6b4d8e9d298a8dd26fd2c1 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9dd888ae6da48b2ba19e55c31761b79 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f7ea7de75df4e09804612ffa49b9b64 : SUCCESS in 3m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38e1d0effd6949ec90c786bb9624007a : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c12229352594add9bf312ba301c4e23 : SUCCESS in 3m 23s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/b0efde6de28843058471e7a6616a833d : SUCCESS in 28m 57s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/a09c8d7236a74854b7317310e3fcc46d : FAILURE in 42m 40s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/e64b1a1d99544896833cbf45dfb1fc02 : FAILURE in 32m 31s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/f3222424e3764a8b80d07aa798f60122 : FAILURE in 41m 49s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/b53c28cd24d742eab120b57a44193889 : FAILURE in 33m 23s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/69b004f8b268436091f6b752b5247b8c : SUCCESS in 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0adbcba584f4372468087539d193b90fbd5a9029","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-10 09:53:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"5731db5bc8411fa3c48d0b07560c889c57bd3ff0","tag":"autogenerated:gerrit:restore","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-02-26 09:33:35.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"75636fbc3fd4cf641d3e59230648dcf90a499bf2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-02-26 09:33:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nThe following files contain Git conflicts:\n* tox.ini\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fb5ca88f50c9ccc5a07d06f9629de91eb968f8ce","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-02-26 09:36:09.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"229440d19bd3cb79c49b5990c014a5433c54d96f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-02-26 09:36:38.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a6534ddbee28973c4d385186ac5569efd591bf0f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-02-26 09:39:19.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"893adba3eb2585ca0cfd80aad7d3a21a8a581d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 10:27:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4be8232bb37d42f5980ceb9d2afe6fbd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78ca6534b0ed487795fffa05e632e352 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92c750e8d5974c14ab92ed49669166fb : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42605e371fb14f73890055ceee8d8a27 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/028e175701a8466d86be02bf42052d23 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0314897e0c644aa78f25a27954f8c815 : FAILURE in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7215713d1b994056898b95866619ec9f : SUCCESS in 3m 34s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/34877e97459548db9feacff29220494c : SUCCESS in 33m 14s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/ee542602a6284822bf6af7809564d25d : FAILURE in 33m 10s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/4d95494f3927481386ad9f3e0ae79dbe : FAILURE in 37m 58s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/27e1bcd39ff94a42a08a4641d7ee6fed : FAILURE in 34m 31s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/c345979ca53848fead7908c2f3149275 : FAILURE in 42m 16s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/5d974fa204c748248bed36bf6be0ae6b : FAILURE in 35m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c043478309569ec820c8860b9e4e4b32e6d302a0","revisions":{"5e303b3d90fcf7fe73b3ad08e76482af37b36eca":{"kind":"REWORK","_number":1,"created":"2022-08-06 09:06:23.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/33/852333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/33/852333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/1"}}},"commit":{"parents":[{"commit":"e659f67b992408f08e7b0900094bfc4e3a948f21","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/e659f67b992408f08e7b0900094bfc4e3a948f21"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-06 09:03:38.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-06 09:03:48.000000000","tz":480},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: Ibc07a8f3ecde14377a1afe33adb4d41307a00b2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/5e303b3d90fcf7fe73b3ad08e76482af37b36eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/5e303b3d90fcf7fe73b3ad08e76482af37b36eca"}]},"branch":"refs/heads/master"},"e8511ab76ca7a3d91a7046a08d25cea4cda25bd1":{"kind":"REWORK","_number":2,"created":"2024-02-26 09:33:56.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/33/852333/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/33/852333/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/2"}}},"commit":{"parents":[{"commit":"607f7a0716ca574b8d9a83264c07910e15158b8a","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/607f7a0716ca574b8d9a83264c07910e15158b8a"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-06 09:03:38.000000000","tz":480},"committer":{"name":"Ke Niu","email":"niuke@inspur.com","date":"2024-02-26 09:33:56.000000000","tz":0},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: Ibc07a8f3ecde14377a1afe33adb4d41307a00b2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/e8511ab76ca7a3d91a7046a08d25cea4cda25bd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/e8511ab76ca7a3d91a7046a08d25cea4cda25bd1"}]},"branch":"refs/heads/master","description":"Rebase"},"01d07d6c1fcb0bfa963459f05e9ea8dcc25ac9b7":{"kind":"REWORK","_number":3,"created":"2024-02-26 09:36:09.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/33/852333/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/33/852333/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/3"}}},"commit":{"parents":[{"commit":"607f7a0716ca574b8d9a83264c07910e15158b8a","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/607f7a0716ca574b8d9a83264c07910e15158b8a"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-06 09:03:38.000000000","tz":480},"committer":{"name":"Ke Niu","email":"niuke@inspur.com","date":"2024-02-26 09:36:08.000000000","tz":0},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: Ibc07a8f3ecde14377a1afe33adb4d41307a00b2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/01d07d6c1fcb0bfa963459f05e9ea8dcc25ac9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/01d07d6c1fcb0bfa963459f05e9ea8dcc25ac9b7"}]},"branch":"refs/heads/master"},"c043478309569ec820c8860b9e4e4b32e6d302a0":{"kind":"REWORK","_number":4,"created":"2024-02-26 09:36:38.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/33/852333/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/33/852333/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/852333/4"}}},"commit":{"parents":[{"commit":"607f7a0716ca574b8d9a83264c07910e15158b8a","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/607f7a0716ca574b8d9a83264c07910e15158b8a"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-06 09:03:38.000000000","tz":480},"committer":{"name":"Ke Niu","email":"niuke@inspur.com","date":"2024-02-26 09:36:38.000000000","tz":0},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: Ibc07a8f3ecde14377a1afe33adb4d41307a00b2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/c043478309569ec820c8860b9e4e4b32e6d302a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/c043478309569ec820c8860b9e4e4b32e6d302a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
