)]}'
{"id":"openstack%2Fkuryr-libnetwork~877180","triplet_id":"openstack%2Fkuryr-libnetwork~master~I411086127cfaee20c17867adf779200e6e31a2ea","project":"openstack/kuryr-libnetwork","branch":"master","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2023-12-20 08:15:17.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}},"hashtags":[],"change_id":"I411086127cfaee20c17867adf779200e6e31a2ea","subject":"Remove external cluster store from docker config","status":"MERGED","created":"2023-03-12 06:53:10.000000000","updated":"2023-12-20 09:40:05.000000000","submitted":"2023-12-20 09:35:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877180","meta_rev_id":"de0af27ab55a7a44627d72e91e10e53f0a9957ea","_number":877180,"virtual_id_number":877180,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-20 09:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2023-12-20 05:29:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2023-12-20 05:29:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-12 07:38:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8354af97571e65e31f5fc9c9cda87bbe621a4c72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-03-12 06:53:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c4cc0b6f31a1d84f5ae73792494d285740a43ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-12 07:38:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa162f275147469ca408a54b731d7888\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcbbdb7212b245bfb117ff3e21b1ec7b : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e686a625ca64972b1577bd36de4a383 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7e60b2fb32646719ea5cda81c14f2c3 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c07447ae3c24438b18045d38688583b : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5d55cec36494be4b0281b41b2f14705 : SUCCESS in 3m 36s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/1b9fdc65e2a24b22bb870545dcb1e8e2 : SUCCESS in 19m 08s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/aeccd24568764815aea549a72426854d : SUCCESS in 20m 11s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/9b63c4f6a3544a23ad0932601d1ef3b4 : FAILURE in 39m 59s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/56f387b2dbf24688bd5f1e590eb933fe : SUCCESS in 18m 56s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/9399f785bdf348629b714aca02a29386 : FAILURE in 22m 29s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/f76cba8b9a384645bd7594e9c8a4eb78 : SUCCESS in 20m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7829d649d9d58893da94b850253a6dca031e344c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-03-12 07:49:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4be5dfbe081f4c24bf4a4f754f94c68830d4485d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-12 08:18:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c02f4e203da54a3e8136236bfa94a181\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9b0c09a778e4a9688b7814ea2ad8d7b : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b15992209a0a49f7b2afadace0c55f36 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fad33af027634623b96f40457472e1e4 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/301213ec4458468da45839b9c798f655 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f243dedb584fb58db907c66aa81d5f : SUCCESS in 3m 08s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/e75774b28f7f4d6087e8c8569538ab5d : SUCCESS in 18m 55s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/b75fa2accf8246b48d52248213fc6ce5 : SUCCESS in 25m 40s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/a5be0cba7c65464d83f52132aee64251 : FAILURE in 26m 49s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/e4eefd3793534a7ca99019d062c647c1 : SUCCESS in 23m 40s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/93fffeb03f5843d5b9553ffb8b691e41 : FAILURE in 26m 31s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/03d6c0d6b30543598976dd6d87bfed0c : SUCCESS in 25m 42s","accounts_in_message":[],"_revision_number":2},{"id":"f8e767b286831bea7c2ae832d11a6396aee99efc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-12-18 12:33:58.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7a0e83fa448d55735cdee622d2fe68d05418230d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 13:23:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a01e756da70a44929b1dca30cca40769\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3379ed4be04405ea9adf588b30965b9 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4719941bb7d4e388c36b42def8ec253 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c548742646743eead51eb05975bbdc1 : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83b48ce8cd364da3b873d6df88b48401 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19b02dda4aed420b995c94b6fa5047e1 : SUCCESS in 5m 30s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/d111a47e5d724d92b5ebe6f9b5d78424 : SUCCESS in 22m 15s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/729269e145864a03ac9af3ec4c23c858 : SUCCESS in 26m 59s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/5d9d8c1a2bee468b8aceb11840c0f7e9 : FAILURE in 38m 50s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/a93e70143ccf4ac8aa1232ac9a5262fc : FAILURE in 42m 33s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/29aef58521fa4213b778d83db359669e : FAILURE in 24m 31s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/099d6674dd48427b845c250f78cf2355 : FAILURE in 23m 05s","accounts_in_message":[],"_revision_number":3},{"id":"afb5f35c3d44ac2b2d1f60754f267a9e42d12cc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-12-19 05:00:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5ec136bcae37d864aff1d9db4ff95a99c6d1424d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-12-19 05:42:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8d4f0f019e5287bae034e84112e08bbba4902e48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 05:46:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eaaa2ce4f144b3282a178e896d2cc22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb2eb63b7f4d4081bef9fc0800d01908 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92e51d60ac904a26b0754c77249f6088 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e40f16542364368ab260bec85285123 : SUCCESS in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e0e6c5caf274cd5baef5de75c592005 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aedb6ce2f26c458b9ad27c257f881001 : SUCCESS in 6m 11s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/cb2774c6527e4c5fa780989e0ed64a70 : SUCCESS in 22m 46s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/1da5b6c6ca80416f9654b0adf032d0ee : SUCCESS in 26m 45s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/b865dd1ff3c743bca5b94d55ebe51dd8 : FAILURE in 29m 54s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/8d39ee6269db4ee884e1a919fff04313 : SUCCESS in 21m 25s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/cb40043d7bd048f0b73801d945ca7e10 : FAILURE in 39m 35s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/a6a1b67c9aef40a69b8b9bb0dc1f1f5b : SUCCESS in 20m 45s","accounts_in_message":[],"_revision_number":4},{"id":"c8f256c0f9219fb76e486e4531287fe59f37b03a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 05:46:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8de9f18b101e3dafd2ba9688fed913e14a2f68f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 06:29:14.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/189445a272944f5b9b7379a5bb8ef3dd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42933f6644c14ebc91d38e8752df8030 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/750b6d8276a84b9a91894bfc25ae3c20 : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e46c35b6f3724578a302b42cd2e48752 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c85b6e69dd41ccbc35647541877b53 : SUCCESS in 3m 20s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/4f5e6be6b58f4b18b610af36223c8996 : FAILURE in 35m 25s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/383e157b9cf94ecabf42f9a0b97c79d9 : SUCCESS in 34m 55s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/b6aca4223a6c4a92bc166a4d4e237afc : FAILURE in 21m 43s","accounts_in_message":[],"_revision_number":4},{"id":"ae14e737702d2591d1d3bc57ea05c7e24eee64d1","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-12-19 07:26:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d7061f7e964b7ddbe2a7006bc71fed1d2bbf2e53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 08:09:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ab5827f8c7d4e3a94de203ccb1ee529\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abb68803105c43379248526c48bce963 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fa4a360947e4df19573396fe1baaabe : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb0a6db215a84dd4885ecdba3bec6a2e : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/268d5b7a9a5b448ebb97fc6b36164d0e : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/596adaa74fbf4c75ba1f438dffea2a40 : SUCCESS in 3m 27s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/7c5e3da2185547ca8d316873e27ccba7 : SUCCESS in 16m 26s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/f691b60f6e5441519a4c5e4a017c3c0d : SUCCESS in 25m 56s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/f4e3d088e3a9429ca4213f872f4cd48f : FAILURE in 19m 48s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/b28f787bbf3b43ad9751161b96bcd5d8 : SUCCESS in 35m 27s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/b4a1404a2dc34ee694cb8cd3883d08e5 : FAILURE in 36m 40s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/2745705f01ab47e19e8c81ad504454e2 : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":4},{"id":"c96ad32fe0152f3c56ce68efc41740c7ec7f3ebf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 08:10:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c8a5f22c08ff0489a06c499f873b9b05675231b0","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-12-19 08:13:02.000000000","message":"Patch Set 4: -Code-Review Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"52f5af7d6fdbfb9cacd2b81edf6017d6e3b1b3c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 08:13:15.000000000","message":"Patch Set 4: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":4},{"id":"8a0b7c15d338264063000d78878aa76adbefaff9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-12-20 05:29:38.000000000","message":"Patch Set 4: -Workflow Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"10d3033189b995ba08c2a07061f828b4dfada4fc","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-12-20 05:29:42.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ce6708fa829f455a23ed1d2ec4b5b2a9f58a3b1c","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-12-20 05:44:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"37f4d70a6c2c19f255b307db94e5089c93bb4c92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 06:13:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e84df45490984a909459d15db98cc8eb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc164134914b4311b8cc53e2f6646611 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90d0f0cb792740e0b583a658adc39194 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dcca460ebe64466bce2860a56b9e610 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de31aa755d19443983daa69e7902de61 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dc12edaffd545ad9cfbe2774497daaf : SUCCESS in 3m 19s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/e2d506345e6846cfb21277d04cac4ff1 : SUCCESS in 32m 13s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/5f7ad46129964f099ffb7eadc30b6a9f : SUCCESS in 37m 04s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/7fda20c6e2a545cf8533e094999b16d7 : FAILURE in 28m 47s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/94e7917bdaef43e98394fca46853f20f : SUCCESS in 34m 31s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/a9f2173d98fb4ab1ba658636464e07cd : FAILURE in 29m 20s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/71d69dd9f90843769c9d79a8af8d3586 : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":4},{"id":"ce6bbed9dfe5e34d399724d4bf8eb544985f6dc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 06:13:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b56ffa43150ca086411746ea1c0d0aad8855665d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 07:03:13.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a17ead4a1624491dbf4ac7eeb29db0ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca2668fdae14505980fa94ffce34fb7 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d85d30660d594d5fa6a2663c2f9e121f : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ce55ddc7afa47aca967060a7774ed6c : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d1f25fd41254ffba2b9e3e3706a136a : SUCCESS in 3m 17s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/140e60794e1b40feaf79210e02a8acb1 : SUCCESS in 27m 55s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/92cc0c79332d4f36895555d38507ee5d : FAILURE in 43m 33s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/a7cbf8b9bfcc430da00e08f4d2498811 : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":4},{"id":"6448ff329d8130c662778d0653063b580a82c19d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2023-12-20 08:15:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b6527c87899cfacd30386cba886236995b8f49c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 09:05:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ded644817c4c4d93b54f9bfc2d740172\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfd92884096f44a18c5edd97419a283e : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efe779571eff4fb198f722c2cc4fbec5 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc60d8636ee74aeb8f2e6f16d92d54ef : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d91d2394a9774a3f8dd4ade96674bab5 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1949c33adf0d467e87bdc1ae13fe5e03 : SUCCESS in 3m 51s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/2a2a9a4b2fa549f1906c8fec1928a9f3 : SUCCESS in 21m 07s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/6cdbf48b82af4c8e8defa4e58b3bc58f : SUCCESS in 29m 04s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/99d0ac1ebe0b498ebc81903f85ca999d : FAILURE in 24m 03s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/0ea026288b7849e9806c5424650987c5 : SUCCESS in 28m 00s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/3185eeffb6644e558407c1ca9d855776 : FAILURE in 39m 55s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/8b6166e0126c45bc8169d8b590d2962e : SUCCESS in 32m 38s","accounts_in_message":[],"_revision_number":4},{"id":"4ae1b508a8c2ddcf68927f7a44e1801d510a0b9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 09:06:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"883d5b4b13e7364519bf15a0505840e7a4a0de24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 09:35:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a9fa87d45cb45dd85ffb0a683ae6644\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf75fdf559b479995407e7fb650782f : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87f0bbdf8e424eb78292bec1826041ba : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9fc3cbe835b425b80ff89a00cd6ee48 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d44a99373a746a99e8635d20fc36fba : SUCCESS in 4m 56s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/5526be22d7b148b7815720928098d23a : SUCCESS in 25m 21s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/7d3cce777440412d96c59fecfbc6de71 : SUCCESS in 27m 27s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/455d3e82c99648b4a7848fb081f9d063 : SUCCESS in 23m 45s","accounts_in_message":[],"_revision_number":4},{"id":"b24f91d7efb9d42ec4320a20295ae5771b3c82f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 09:35:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"de0af27ab55a7a44627d72e91e10e53f0a9957ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 09:40:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e62d7b679fa47a1a174fc34943aa706\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31043fc9bae54a0cbc0f74f653dd5d7f : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"00ab075497d03e62c9c57adf5d5e1f2210acb61f","revisions":{"484ed96b67295389c512eb284eb4c200aed17b2a":{"kind":"REWORK","_number":1,"created":"2023-03-12 06:53:10.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/80/877180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/80/877180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/1"}}},"commit":{"parents":[{"commit":"3bce4281376615a1a5b3cc347670261155f4ea50","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3bce4281376615a1a5b3cc347670261155f4ea50"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2023-03-12 06:51:43.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2023-03-12 06:52:51.000000000","tz":0},"subject":"[WIP] Remove external cluster store from docker config","message":"[WIP] Remove external cluster store from docker config\n\nChange-Id: I411086127cfaee20c17867adf779200e6e31a2ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/484ed96b67295389c512eb284eb4c200aed17b2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/484ed96b67295389c512eb284eb4c200aed17b2a"}]},"branch":"refs/heads/master"},"f79962f746d7be9647b9a1cbc77df27eb593b046":{"kind":"REWORK","_number":2,"created":"2023-03-12 07:49:51.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/80/877180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/80/877180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/2"}}},"commit":{"parents":[{"commit":"3bce4281376615a1a5b3cc347670261155f4ea50","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3bce4281376615a1a5b3cc347670261155f4ea50"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2023-03-12 06:51:43.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2023-03-12 07:49:39.000000000","tz":0},"subject":"[WIP] Remove external cluster store from docker config","message":"[WIP] Remove external cluster store from docker config\n\nChange-Id: I411086127cfaee20c17867adf779200e6e31a2ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/f79962f746d7be9647b9a1cbc77df27eb593b046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/f79962f746d7be9647b9a1cbc77df27eb593b046"}]},"branch":"refs/heads/master"},"1e4dc6fa9c377542c32896715e01c6d58e3b2c21":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-18 12:33:58.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/80/877180/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/80/877180/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/3"}}},"commit":{"parents":[{"commit":"693ffb903b409abd18f321af05871b64d9cb5084","subject":"Fix python path due to global virtual environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/693ffb903b409abd18f321af05871b64d9cb5084"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2023-03-12 06:51:43.000000000","tz":0},"committer":{"name":"hongbin","email":"hongbin034@gmail.com","date":"2023-12-18 12:33:58.000000000","tz":0},"subject":"[WIP] Remove external cluster store from docker config","message":"[WIP] Remove external cluster store from docker config\n\nChange-Id: I411086127cfaee20c17867adf779200e6e31a2ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/1e4dc6fa9c377542c32896715e01c6d58e3b2c21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/1e4dc6fa9c377542c32896715e01c6d58e3b2c21"}]},"branch":"refs/heads/master","description":"Rebase"},"00ab075497d03e62c9c57adf5d5e1f2210acb61f":{"kind":"REWORK","_number":4,"created":"2023-12-19 05:00:33.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/80/877180/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/80/877180/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/80/877180/4"}}},"commit":{"parents":[{"commit":"693ffb903b409abd18f321af05871b64d9cb5084","subject":"Fix python path due to global virtual environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/693ffb903b409abd18f321af05871b64d9cb5084"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2023-03-12 06:51:43.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2023-12-19 05:00:31.000000000","tz":0},"subject":"Remove external cluster store from docker config","message":"Remove external cluster store from docker config\n\nRelated-Bug: #2007142\nChange-Id: I411086127cfaee20c17867adf779200e6e31a2ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/00ab075497d03e62c9c57adf5d5e1f2210acb61f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/00ab075497d03e62c9c57adf5d5e1f2210acb61f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
