)]}'
{"id":"openstack%2Fkuryr-libnetwork~894965","triplet_id":"openstack%2Fkuryr-libnetwork~master~I73e536927ecd8f0a52d1464e0496766e60aa1758","project":"openstack/kuryr-libnetwork","branch":"master","topic":"release-health-check-cwr-bobcat","hashtags":[],"change_id":"I73e536927ecd8f0a52d1464e0496766e60aa1758","subject":"DNM: gate health test","status":"ABANDONED","created":"2023-09-13 19:18:24.000000000","updated":"2023-09-14 07:12:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"725638db698051bdc04e79563c1e8409a0972eea","_number":894965,"virtual_id_number":894965,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-13 20:06:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-13 20:06:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35ac7ecb6363b88f04aa49f26c8db67057de5662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-13 19:18:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edc066393f90acc5b1b7eaf2e874751a2382dafe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 20:06:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c30f3240f3c47938b276f2c44ce3974\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6305a7bd05a04a8c891740f8784a9837 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63085ab4e0bb411da08c670ac21060a9 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8dc1e83a3484ec798c7798054ee998f : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fda2b8d7b394348b5839a4af428f0d0 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa91b53eb2334c7e839a0efff05cdf0e : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af1c6e89d3b24881a789a817afa3b146 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e028325109ae4a749e81b9d43d0eaa43 : SUCCESS in 6m 26s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/64826065f0f5481497974dd4920466f2 : SUCCESS in 25m 21s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/83a284a0fd014cf6a208a942bba5d257 : SUCCESS in 30m 35s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/1c29e44f709643ebbf50039574359e81 : FAILURE in 34m 02s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/d2ed5f90fc5e456e9dd6b06d16f88d7a : SUCCESS in 27m 23s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/2841c15b4e9746ce97278626cd760227 : FAILURE in 40m 22s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/5d6d3b1d60034efe81153569bca71c25 : SUCCESS in 27m 03s","accounts_in_message":[],"_revision_number":1},{"id":"725638db698051bdc04e79563c1e8409a0972eea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-14 07:12:53.000000000","message":"Abandoned\n\ngate looks healthy, abandoning.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d06942b1cb683e8b9d25abbee35c1c9db01538d","revisions":{"8d06942b1cb683e8b9d25abbee35c1c9db01538d":{"kind":"REWORK","_number":1,"created":"2023-09-13 19:18:24.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/65/894965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/65/894965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/65/894965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/65/894965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/65/894965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/65/894965/1"}}},"commit":{"parents":[{"commit":"3bce4281376615a1a5b3cc347670261155f4ea50","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3bce4281376615a1a5b3cc347670261155f4ea50"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:18:15.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:18:22.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I73e536927ecd8f0a52d1464e0496766e60aa1758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8d06942b1cb683e8b9d25abbee35c1c9db01538d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/8d06942b1cb683e8b9d25abbee35c1c9db01538d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
