)]}'
{"id":"openstack%2Fkuryr-libnetwork~928633","triplet_id":"openstack%2Fkuryr-libnetwork~master~I2c0affe840a81be2c8eda3a09b75f3016afcddbc","project":"openstack/kuryr-libnetwork","branch":"master","topic":"release-health-check-cwrc-dalmatian","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-09-25 15:08:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2c0affe840a81be2c8eda3a09b75f3016afcddbc","subject":"DNM: gate health test","status":"ABANDONED","created":"2024-09-09 15:25:21.000000000","updated":"2024-09-25 15:08:20.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7384428988c9bc6f3bde8284deabde78d75dec94","_number":928633,"virtual_id_number":928633,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-25 11:44:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-09 16:16:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-25 01:45:19.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"CC"}],"messages":[{"id":"9f8e8a134ec35f774a401100d6aa5723950075be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-09 15:25:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95b8a4d443f15b0b261b7842ea5df69212cb1b39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 16:16:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35bd95491ba84304ab6ac0ecd2d3316b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a7e731d8dab4486878f9334989e6832 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efc90fe31d234168bc62c5eee7819150 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64c60317a7d647519e2392ddf7c2b159 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3d3ef082a5d4005b996c31d38c9be35 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57d01cb1a7dc40c883a59a983157fdf3 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b61c1441e3a41aa9549b3ca9eeb1bc7 : SUCCESS in 3m 50s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/e5edce38ee6d47888e808e2bb09178c8 : SUCCESS in 19m 27s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/cbf55b985c974516abef62a5fd1e115f : FAILURE in 30m 50s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/95ae15cd3bc44ac4a8eff962c3fc5d76 : FAILURE in 29m 53s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/3b84e7abceda47378720a18aa64839f0 : FAILURE in 19m 21s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/22df86e7313f4cee8347307e527f4b74 : FAILURE in 20m 01s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/1ff12b91eb81464cbf21de8bb7047b10 : FAILURE in 26m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ac39947f118e1693c1bdb65b15a9a95e40b57510","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-09-25 01:45:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a7e3b861b0f62a9249d36a8b1b668ee805c507c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 02:29:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c0e6338f1204dbf8e53895624055dd3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ca56e143c0346649bbb4afed66d0786 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/641cda55ed0d4f5ab4726431888430bb : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/644a7e56231c407e8063651287bd18d2 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/99ec9df0d39c4ce08bca580e6d633dfc : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ca7f7b187734f4bbc0320776b7add2e : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/babe19bdb788432fb884b69b0bfa34e3 : SUCCESS in 2m 45s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/1d641bcca458445db53332a5ea968ab6 : SUCCESS in 29m 29s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/3d746bf6711d4735a1f37d95fdb6c24f : FAILURE in 16m 58s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/b45731b9585248af97ac6649d406e99d : FAILURE in 33m 25s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/b2caa478f8d949fda31cab0afda9de69 : FAILURE in 31m 44s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/59d8222bda20442f90cc2aa9fb0c3879 : FAILURE in 27m 55s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/2a9642454d6248e592cae71e17348acb : FAILURE in 19m 01s","accounts_in_message":[],"_revision_number":1},{"id":"744005dc2356c23cb4d641c2235f3ebf11c6338b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"real_author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-09-25 11:03:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_17685\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"_revision_number":2},{"id":"35b6f789cf21e2dbc44eaefe26f905e86a434237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 11:44:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1af7816447344a7a347f964a569f6ef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35cf52d127ad4b9ebc7e7d30e7f92614 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e85e6d6a33354c55bf80327bcc16cff8 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da11c06e8d724844aa7f3ff1e5d71d09 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23d724e82550482e97226775e90aaf62 : SUCCESS in 5m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b305e8fb44f8424ab4f6012b88c4636a : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/644ab8d525cc46aba65ca8d7c7438c8b : SUCCESS in 3m 01s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/c173ce7f32d249efbae34e642030283e : SUCCESS in 27m 42s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/c36ccd2923f441bb981b1b731874803d : SUCCESS in 29m 18s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/fcf77a2597be46bcbbd05aff3841c026 : FAILURE in 33m 59s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/cb16bcce55e347d1a0d780bb75862eba : SUCCESS in 32m 48s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/4238092f3ac44a5c94fde69d4aca3454 : FAILURE in 20m 57s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/1991c7dea81a4bd3b0f7edfd7cb36a0b : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":2},{"id":"7384428988c9bc6f3bde8284deabde78d75dec94","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-25 15:08:20.000000000","message":"Abandoned\n\ngate is unblocked, abandoning.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"482b37109b289673e0b86b2bf2bac81a262a5d83","revisions":{"9348e6dc0730aaf8a0148de3f7a610e2d39d0694":{"kind":"REWORK","_number":1,"created":"2024-09-09 15:25:21.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/33/928633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/33/928633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/928633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/928633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/928633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/928633/1"}}},"commit":{"parents":[{"commit":"c2fc51c2222aa893ec9504e7335c6203a1f20b63","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/c2fc51c2222aa893ec9504e7335c6203a1f20b63"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-09 15:25:06.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-09 15:25:19.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I2c0affe840a81be2c8eda3a09b75f3016afcddbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/9348e6dc0730aaf8a0148de3f7a610e2d39d0694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/9348e6dc0730aaf8a0148de3f7a610e2d39d0694"}]},"branch":"refs/heads/master"},"482b37109b289673e0b86b2bf2bac81a262a5d83":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-25 11:03:09.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"real_uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/33/928633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/33/928633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/928633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/928633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/928633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/33/928633/2"}}},"commit":{"parents":[{"commit":"a4b276cfff1b2d88a58fd6c6a1c99e3761984ebd","subject":"Remove test_create_network_with_same_name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/a4b276cfff1b2d88a58fd6c6a1c99e3761984ebd"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-09 15:25:06.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-25 11:03:09.000000000","tz":0},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I2c0affe840a81be2c8eda3a09b75f3016afcddbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/482b37109b289673e0b86b2bf2bac81a262a5d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/482b37109b289673e0b86b2bf2bac81a262a5d83"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
