)]}'
{"id":"openstack%2Fkuryr-libnetwork~953499","triplet_id":"openstack%2Fkuryr-libnetwork~master~I82399d3b9b6915e8126b77d1f5f71372c48ba4ed","project":"openstack/kuryr-libnetwork","branch":"master","topic":"remove-USE_PYTHON3","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-06-27 09:32:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I82399d3b9b6915e8126b77d1f5f71372c48ba4ed","subject":"Zuul: do not use USE_PYTHON3","status":"MERGED","created":"2025-06-27 00:38:20.000000000","updated":"2025-06-27 09:33:37.000000000","submitted":"2025-06-27 09:32:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953499-remove-USE_PYTHON3","meta_rev_id":"1fa2fd5d3beef2e4f825a925e12fd32138aba699","_number":953499,"virtual_id_number":953499,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-27 09:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2025-06-27 09:00:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2025-06-27 09:00:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-27 01:07:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-27 09:00:23.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"ee1e649056fba8601db05300a19b171bad056488","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-27 00:38:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4659a618ef2739bdbcf8286802a1fb5fb0ed2bba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 01:07:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/959d80a8f9864694bc57e5f940dcbf60\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6762aba731304836911f1a07b527304f : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5171d41947c549fba8db66056a827f25 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/348e53a63e05492ebe8a5b22b11c1a4c : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b94acc4c4f54bf08bca311e6676f310 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eef9465f14694fa99091481c7b8fa4c8 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a1edcb63c954f69921d59650c6d03c4 : SUCCESS in 4m 01s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/40895e502f434b64bc89b1436bf87933 : SUCCESS in 10m 19s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/8b92e117ee6e4132a2dd0f7381790cc1 : SUCCESS in 27m 35s (non-voting)\n- kuryr-libnetwork-kuryr_lib-src-rally https://zuul.opendev.org/t/openstack/build/b1b36f8cbdb84d1c92110f5a3f348178 : FAILURE in 19m 58s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/0b3f0968e2ad4ee0be55b9039edbf16a : SUCCESS in 22m 39s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/c980c26da654424da57b423ebfb07a7c : FAILURE in 27m 11s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/f4418378da2348c9b195406fac71565c : SUCCESS in 15m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e4f8c328a4dc6075ace1b9fc74093139988d9a09","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2025-06-27 09:00:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1669f5711ed1aededfe61a929bd222e21080553a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 09:00:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef1e2e66889c5accfb096c25f9f4d20e68fea037","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 09:32:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f26326e856d41a09bb70a995752c993\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcb2c1a48fe543879118637d81f50e52 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cfc1a665670457fb40c907d32c9527d : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd12e7e68bae47be82f872e1cd21a8fc : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f34d6a4d085d42c7aa2d949e627cd5ae : SUCCESS in 2m 28s\n- kuryr-libnetwork-kuryr_lib-src-fullstack https://zuul.opendev.org/t/openstack/build/3399534d4fd84be4aceec189916fa8d6 : SUCCESS in 27m 00s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/c823cf5f0e4141db831879229491ce5e : SUCCESS in 22m 28s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/1cfecc42b1dc4b3890642bbfbf7c00a1 : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3745c6b133de3b6da2600fffa1141eef7d990b98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 09:32:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1fa2fd5d3beef2e4f825a925e12fd32138aba699","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 09:33:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff457995e5cc4558aab4769faab98f0e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bee4ab9555c44299958ac2c229443b7 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2faa110a5bc1db53d3f96ae239fe934c0e41567d","revisions":{"2faa110a5bc1db53d3f96ae239fe934c0e41567d":{"kind":"REWORK","_number":1,"created":"2025-06-27 00:38:20.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/99/953499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/99/953499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/99/953499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/99/953499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/99/953499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/99/953499/1"}}},"commit":{"parents":[{"commit":"70676593993a8c370be3fc27a8fecc4e84661e24","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/70676593993a8c370be3fc27a8fecc4e84661e24"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-06-27 00:38:05.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-06-27 00:38:18.000000000","tz":120},"subject":"Zuul: do not use USE_PYTHON3","message":"Zuul: do not use USE_PYTHON3\n\nDevstack has removed the USE_PYTHON3 variable[1][2] and now always uses\nPython 3.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/920658\n[2] Commit 5412dbfe7b797149f1f68100de8003b1876398fe\n\nChange-Id: I82399d3b9b6915e8126b77d1f5f71372c48ba4ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/2faa110a5bc1db53d3f96ae239fe934c0e41567d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/2faa110a5bc1db53d3f96ae239fe934c0e41567d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
