)]}'
{"id":"openstack%2Fkuryr-libnetwork~987466","triplet_id":"openstack%2Fkuryr-libnetwork~master~I74c795a61aff5afc40416dc4267905b9453bf50b","project":"openstack/kuryr-libnetwork","branch":"master","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-06-18 02:59:46.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-18 03:32:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I74c795a61aff5afc40416dc4267905b9453bf50b","subject":"Remove url tags from README","status":"MERGED","created":"2026-05-06 09:25:51.000000000","updated":"2026-06-18 03:33:13.000000000","submitted":"2026-06-18 03:32:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987466","meta_rev_id":"28887f50be7e09f2c64a130256ffe65904597d44","_number":987466,"virtual_id_number":987466,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-18 03:32:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2026-06-18 02:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2026-06-18 02:59:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-06 09:35:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 02:24:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"37d5d3740f3d726649b42c43bc065f288ee9bf3c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 09:25:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e43bd3fca4f84ea7bdb2d3c3547eb459696d53f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 09:28:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7fc470e93d9fbc820e4b970f420595c98f989fbc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 09:28:57.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"032b2e4dbefeb10c458f6163725b56d2c9427c14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 09:30:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"22fca71783ea735252b30946634cd7451b6d0d4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 09:35:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abfa70897664449583dad9611319f34c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a3e47cd0904b469f91716306534a7989 : SUCCESS in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb0325f287f149e99100248fe2d39501 : FAILURE in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76768f5973b3494c8b5bbf85161b7c35 : SUCCESS in 3m 23s\n\nWarning:\n  Comments left for invalid file kuryr_libnetwork/opts.py","accounts_in_message":[],"_revision_number":3},{"id":"9a03292954fec6b5a9e1da1b21066311fadc5b58","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 09:46:25.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"ee31af6cc8174ea4e4b0960f4f3e1f1fbf9f314f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 10:56:09.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"940c948c12ec831152ac1eed5703205554fa28cf","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 10:56:11.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"45af7d36432549a04f2df4bc0b3d916733cd84a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 11:10:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22d91e7b4cc14e9a914589d506d2f456\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7d269d3a4c7e40f285a105d42def479b : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2cb92edb0854834bc0199c2795fc4b2 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9d61f53f2644ac1a0e160979ce88674 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":4},{"id":"15ee1d797a83732793d7fe95dea981026bb0bf38","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2026-06-18 02:59:46.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cbb373e61936bf6073021430eeba7bd4f7f13eb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 03:00:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c52ed9817ecae10234b2d5f9f49f0fe239c3ff85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 03:32:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d9216fcae514088885bd85028a7baeb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a11a0f43c8504e519a747fff39f459c5 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36dbb7e1450142479b098522476ae412 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2c8bf4337a54b4293570e21f7fa306d : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5bc9bb786e848149683866f38a15bcf : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":4},{"id":"7a5ff51ce25d5fb0ecb5c1363457cca9c423dac8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 03:32:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"28887f50be7e09f2c64a130256ffe65904597d44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 03:33:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c17d022991a4815b985e5a73f4f2bc4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a980d2f99ff24c96a11c16cdfd17b2ce : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2a0124a27cc155c46eb898dee15d1a0bd8b84276","revisions":{"396b27b68b8b2d82f9740ec4004df3da9c869bd7":{"kind":"REWORK","_number":1,"created":"2026-05-06 09:25:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/66/987466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/66/987466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/1"}}},"commit":{"parents":[{"commit":"3b326567ddbd4253a0f719395328e568c98e1f26","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3b326567ddbd4253a0f719395328e568c98e1f26"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 09:25:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 09:25:51.000000000","tz":0},"subject":"Remove url tags from README","message":"Remove url tags from README\n\nThe tags framework has been discontinued for a long time.\n\nhttps://governance.openstack.org/tc/reference/tags/\n\nhttps://governance.openstack.org/tc/resolutions/20211224-tags-framework-removal.html\n\nChange-Id: I74c795a61aff5afc40416dc4267905b9453bf50b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/396b27b68b8b2d82f9740ec4004df3da9c869bd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/396b27b68b8b2d82f9740ec4004df3da9c869bd7"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"44a7f2b4b7d4f3c621e742be7a81cdf761cb0767":{"kind":"REWORK","_number":2,"created":"2026-05-06 09:28:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/66/987466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/66/987466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/2"}}},"commit":{"parents":[{"commit":"3b326567ddbd4253a0f719395328e568c98e1f26","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3b326567ddbd4253a0f719395328e568c98e1f26"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 09:25:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 09:28:51.000000000","tz":0},"subject":"Remove url tags from README","message":"Remove url tags from README\n\nThe tags framework has been discontinued for a long time.\n\nhttps://governance.openstack.org/tc/reference/tags/\n\nhttps://governance.openstack.org/tc/resolutions/20211224-tags-framework-removal.html\n\nChange-Id: I74c795a61aff5afc40416dc4267905b9453bf50b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/44a7f2b4b7d4f3c621e742be7a81cdf761cb0767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/44a7f2b4b7d4f3c621e742be7a81cdf761cb0767"}]},"branch":"refs/heads/master"},"1855a4e0b058dae0aa664330555b45191e7d9f19":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-05-06 09:30:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/66/987466/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/66/987466/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/3"}}},"commit":{"parents":[{"commit":"cb9eff86d0111354d01787a0aa02e3b27ea83db2","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/cb9eff86d0111354d01787a0aa02e3b27ea83db2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 09:25:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 09:30:00.000000000","tz":0},"subject":"Remove url tags from README","message":"Remove url tags from README\n\nThe tags framework has been discontinued for a long time.\n\nhttps://governance.openstack.org/tc/reference/tags/\n\nhttps://governance.openstack.org/tc/resolutions/20211224-tags-framework-removal.html\n\nChange-Id: I74c795a61aff5afc40416dc4267905b9453bf50b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/1855a4e0b058dae0aa664330555b45191e7d9f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/1855a4e0b058dae0aa664330555b45191e7d9f19"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"44a7f2b4b7d4f3c621e742be7a81cdf761cb0767","theirs":"cb9eff86d0111354d01787a0aa02e3b27ea83db2","contains_conflicts":false}},"2a0124a27cc155c46eb898dee15d1a0bd8b84276":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-05-06 10:56:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/66/987466/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/66/987466/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/66/987466/4"}}},"commit":{"parents":[{"commit":"c0827e8d453a6dfdfa4c707f0bfadf5bfc9d8681","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/c0827e8d453a6dfdfa4c707f0bfadf5bfc9d8681"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 09:25:51.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 10:56:09.000000000","tz":0},"subject":"Remove url tags from README","message":"Remove url tags from README\n\nThe tags framework has been discontinued for a long time.\n\nhttps://governance.openstack.org/tc/reference/tags/\n\nhttps://governance.openstack.org/tc/resolutions/20211224-tags-framework-removal.html\n\nChange-Id: I74c795a61aff5afc40416dc4267905b9453bf50b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/2a0124a27cc155c46eb898dee15d1a0bd8b84276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/2a0124a27cc155c46eb898dee15d1a0bd8b84276"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"1855a4e0b058dae0aa664330555b45191e7d9f19","theirs":"c0827e8d453a6dfdfa4c707f0bfadf5bfc9d8681","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
