)]}'
{"id":"openstack%2Fkuryr-libnetwork~987483","triplet_id":"openstack%2Fkuryr-libnetwork~stable%2F2025.1~Ia2463e7464b28e490c457feb78e681a49856e8be","project":"openstack/kuryr-libnetwork","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-16 15:54:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia2463e7464b28e490c457feb78e681a49856e8be","subject":"Bump hacking","status":"MERGED","created":"2026-05-06 10:57:46.000000000","updated":"2026-05-16 15:55:04.000000000","submitted":"2026-05-16 15:54:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987483","cherry_pick_of_change":987482,"cherry_pick_of_patch_set":1,"meta_rev_id":"73d3ec2ddcf37c1ab94510f2a287bfe2703f2264","_number":987483,"virtual_id_number":987483,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-16 15:54:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2026-05-16 15:11:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2026-05-16 15:11:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-06 11:38:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-16 15:11:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"f6350a88e7e47aeea93bdb3deba49cf2ce0d7e22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 10:57:46.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"350524a368a30d77d773e4c47fea71652fd881d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 11:38:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ce8b90eb18d49e881539a4afd05929b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6e90f0d979b49feb7cc7c13b2b36c81 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0145430c2e5944098a278b19f411e0af : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8b068ad7f834ec0bbd3c459f34ef4b3 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9767fc3f213948d785437d50a77dde77 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8c091e0bf4f4acf952e7c0e53854398 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54eb893a69634cbfadd600c49180fa38 : SUCCESS in 5m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac63c74b2eee425d9156da186879a884 : SUCCESS in 2m 57s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/11ef3f42d0be40178e2d34035672eaad : SUCCESS in 12m 40s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/015a4181494244bca3ebed8d4ade5588 : SUCCESS in 14m 37s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/f0aa729fb21e47ea916ae6226c272835 : FAILURE in 28m 34s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/eafddd9453cc4cfb8ef1b47faefa0041 : SUCCESS in 29m 46s","accounts_in_message":[],"_revision_number":1},{"id":"cb67295f88da8f69237961014dfd296f4439b665","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2026-05-16 15:11:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bfdfec18c4ef70e66a9e9c99e6b537f258bc62e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 15:12:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7b698fa4443bde84b79ee267ca4135f52255e7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 15:54:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/036e1a9bf7f3442e92bbaf3c69698f76\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62c25997d6534d47b8fa3748c3ec9dac : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/507daf7160c54438a9dc8554a8565ce0 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49031a10ad5d4a0fa1c63400f11396e3 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6c95a98671a4de9a9cb77ea856fb1ce : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ad52dc0a0b4a3397b1afd109b4c1f3 : SUCCESS in 3m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0b9d8d4692fc46e58edcbd630cb67a3f : SUCCESS in 2m 15s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/65f7d8c15e65453abd1e95921ca87ed2 : SUCCESS in 29m 46s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/3165d335aa3e46edb335c7f413d3b68e : SUCCESS in 33m 25s","accounts_in_message":[],"_revision_number":1},{"id":"977d3cab577348a68163f993c3de4422ca86a34a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 15:54:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"73d3ec2ddcf37c1ab94510f2a287bfe2703f2264","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 15:55:04.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/751071c46b484a49a1f60fd224999198\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d24143f03a14b6e847ccf9019df0079 : FAILURE in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6172944cc3424e6db1e52cca8816c935 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a964d202278ce2ed9df22f031a67d70bce29c24f","revisions":{"a964d202278ce2ed9df22f031a67d70bce29c24f":{"kind":"REWORK","_number":1,"created":"2026-05-06 10:57:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/83/987483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/83/987483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/83/987483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/83/987483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/83/987483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/83/987483/1"}}},"commit":{"parents":[{"commit":"2d32a1ce080a42b7a6100288918540293c3a21a4","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/2d32a1ce080a42b7a6100288918540293c3a21a4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 09:16:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 10:57:46.000000000","tz":0},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old, new version fix problems with flake8.\n\nChange-Id: Ia2463e7464b28e490c457feb78e681a49856e8be\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/a964d202278ce2ed9df22f031a67d70bce29c24f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/a964d202278ce2ed9df22f031a67d70bce29c24f"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"2d32a1ce080a42b7a6100288918540293c3a21a4","theirs":"cd8ba0164b1e8786567efe00681696d86bcfdeab","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
