)]}'
{"id":"openstack%2Fkuryr-libnetwork~987485","triplet_id":"openstack%2Fkuryr-libnetwork~stable%2F2026.1~I85d45e80819d977e8fea76b95d96e3bbe4062776","project":"openstack/kuryr-libnetwork","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-06-18 02:59:25.000000000","reason":"\u003cGERRIT_ACCOUNT_11536\u003e replied on the change","reason_account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-18 03:32:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I85d45e80819d977e8fea76b95d96e3bbe4062776","subject":"Drop support for Python 3.8 and 3.9","status":"MERGED","created":"2026-05-06 10:58:14.000000000","updated":"2026-06-18 03:33:23.000000000","submitted":"2026-06-18 03:32:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987485","cherry_pick_of_change":987463,"cherry_pick_of_patch_set":4,"meta_rev_id":"dbbc5fd5630836d26c440e878072b955e4480200","_number":987485,"virtual_id_number":987485,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-18 03:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2026-06-18 02:59:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2026-06-18 02:59:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-06 11:43:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 02:25:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"9cbf03a45e71bfd719088e6c0a8678705fea29b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 10:58:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"acdd80cb0bde2e409ae7539c7338d043c5ba340b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 10:58:34.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"26fdc40308e4f18b1454e788d524dfa2601ffbaf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 10:59:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7dd6ddf013148d0b2ec84563e12bc642db93de95","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 10:59:39.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5f2e8ba64bdda2bd5d4c86acb2a20faa4056001a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 10:59:42.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"8ca437c2528993e1bd57aec5358a4e7d7099a188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 11:43:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce3d85b3b7644f5d95f050da0c7f125d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3a990ae6037b4dc99e9c7867d6cc6ffd : SUCCESS in 2m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6135b8bb2a604ffbb68cd144b22819e9 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/186c893082b041cc80f86fe02421a550 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/283cbd18340243cab8c31922bcc5bd13 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7dd2bab5ae9344d59c3a9502cfbb5108 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36684f8f591248398cf0abee8a0508f9 : SUCCESS in 4m 09s\n- kuryr-libnetwork-install https://zuul.opendev.org/t/openstack/build/cfb633259a064b8c8c98066fbd8233e0 : SUCCESS in 17m 43s (non-voting)\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/a67a18eda1a24be4897a0da32f6a9604 : SUCCESS in 30m 44s\n- kuryr-libnetwork-rally https://zuul.opendev.org/t/openstack/build/7b724a7d3f214913aea8c424d86ad087 : FAILURE in 14m 01s (non-voting)\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/9c81d66255d140d58189962cb8f2747e : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":3},{"id":"7d90a7473b35ae15a01e7d63e401ee578c7a3ff0","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2026-06-18 02:59:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"44b2910a9237bd2537c929fc58c38a1a95b046a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 02:59:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"99097f884fc2ae52fa93acbbb955a5aaa42c60d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 03:32:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ad7176d774345bf8c713d38a3b6c2d2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/71ce5254f3c24e64b766b182c2f47c4f : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a768b4984d44339a42a12eabda5fac3 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/137c885d039a422283600282c65ea5be : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1af396df08e424eb02f9877c94488c1 : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c59d467e8bb4cb5b33fc7ade8861618 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aa822d76b0f4a688258102d65d4a7c1 : SUCCESS in 5m 14s\n- kuryr-libnetwork-fullstack-py3 https://zuul.opendev.org/t/openstack/build/a166ca968836477c9973567ba75bcb06 : SUCCESS in 32m 00s\n- kuryr-libnetwork-fullstack-ipv6-only https://zuul.opendev.org/t/openstack/build/4154974208444547a053373731065101 : SUCCESS in 16m 22s","accounts_in_message":[],"_revision_number":3},{"id":"ef569ea11120fe5831ab755eaf15ff51271b6537","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 03:32:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"dbbc5fd5630836d26c440e878072b955e4480200","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 03:33:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7e678e1768640cfbbf19e8fca7e26b8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/184067dd3af64f7e8fa20ec8f718b409 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"87ad0664b07f3f9f214c8ca3389fb75c6bad0c4b","revisions":{"351f077ab25551dd429bc63694c3330729bfe3de":{"kind":"REWORK","_number":1,"created":"2026-05-06 10:58:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/987485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/85/987485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/85/987485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/85/987485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/85/987485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/85/987485/1"}}},"commit":{"parents":[{"commit":"3b326567ddbd4253a0f719395328e568c98e1f26","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/3b326567ddbd4253a0f719395328e568c98e1f26"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 09:18:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 10:58:14.000000000","tz":0},"subject":"Drop support for Python 3.8, 3.9 and 3.10","message":"Drop support for Python 3.8, 3.9 and 3.10\n\nAs per latest testing runtime, we stopped testing the\npython 3.8, 3.9 and 3.10 so dropping their support.\n\nChange-Id: I85d45e80819d977e8fea76b95d96e3bbe4062776\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/351f077ab25551dd429bc63694c3330729bfe3de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/351f077ab25551dd429bc63694c3330729bfe3de"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"ours":"3b326567ddbd4253a0f719395328e568c98e1f26","theirs":"135587f6e5901789ca403f8e84830e032b301cf0","contains_conflicts":false}},"14bf245e493c7d73878c83683eedbf08edb33d62":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-06 10:59:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/987485/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/85/987485/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/85/987485/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/85/987485/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/85/987485/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/85/987485/2"}}},"commit":{"parents":[{"commit":"a030c8f9b9dca4b1400c5cc8eeb22bf47d610ae3","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/a030c8f9b9dca4b1400c5cc8eeb22bf47d610ae3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 09:18:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 10:59:04.000000000","tz":0},"subject":"Drop support for Python 3.8, 3.9 and 3.10","message":"Drop support for Python 3.8, 3.9 and 3.10\n\nAs per latest testing runtime, we stopped testing the\npython 3.8, 3.9 and 3.10 so dropping their support.\n\nChange-Id: I85d45e80819d977e8fea76b95d96e3bbe4062776\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/14bf245e493c7d73878c83683eedbf08edb33d62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/14bf245e493c7d73878c83683eedbf08edb33d62"}]},"branch":"refs/heads/stable/2026.1","description":"Rebase","conflicts":{"ours":"351f077ab25551dd429bc63694c3330729bfe3de","theirs":"a030c8f9b9dca4b1400c5cc8eeb22bf47d610ae3","contains_conflicts":false}},"87ad0664b07f3f9f214c8ca3389fb75c6bad0c4b":{"kind":"REWORK","_number":3,"created":"2026-05-06 10:59:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/987485/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-libnetwork","ref":"refs/changes/85/987485/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/85/987485/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/85/987485/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/85/987485/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-libnetwork refs/changes/85/987485/3"}}},"commit":{"parents":[{"commit":"a030c8f9b9dca4b1400c5cc8eeb22bf47d610ae3","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/a030c8f9b9dca4b1400c5cc8eeb22bf47d610ae3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 09:18:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 10:59:39.000000000","tz":0},"subject":"Drop support for Python 3.8 and 3.9","message":"Drop support for Python 3.8 and 3.9\n\nAs per latest testing runtime, we stopped testing the\npython 3.8 and 3.9 so dropping their support.\n\nChange-Id: I85d45e80819d977e8fea76b95d96e3bbe4062776\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/87ad0664b07f3f9f214c8ca3389fb75c6bad0c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-libnetwork/commit/87ad0664b07f3f9f214c8ca3389fb75c6bad0c4b"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
