)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~618751","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I40bf39c4c1d0b43f5d83e6aed81786390753ce99","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"remove-multithread","hashtags":[],"change_id":"I40bf39c4c1d0b43f5d83e6aed81786390753ce99","subject":"Enhance svc namespace isolation for LoadBalancer type svc","status":"MERGED","created":"2018-11-19 13:42:36.000000000","updated":"2019-07-09 15:50:25.000000000","submitted":"2019-07-09 15:50:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"618751-1562687425215-29425bd4","meta_rev_id":"f8e9d62d4c20d8ddf325666a7dac1d6df9c06251","_number":618751,"virtual_id_number":618751,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":2,"date":"2019-07-09 15:50:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":26287,"name":"Yossi Boaron","email":"yboaron@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"all":[{"value":2,"date":"2019-07-09 12:03:40.000000000","_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":2,"date":"2019-07-09 12:03:40.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":1,"date":"2019-07-09 12:03:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":26287,"name":"Yossi Boaron","email":"yboaron@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-09 12:04:44.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":26287,"name":"Yossi Boaron","email":"yboaron@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":26287,"name":"Yossi Boaron","email":"yboaron@redhat.com"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-19 13:43:08.000000000","updated_by":{"_account_id":26287,"name":"Yossi Boaron","email":"yboaron@redhat.com"},"reviewer":{"_account_id":26287,"name":"Yossi Boaron","email":"yboaron@redhat.com"},"state":"REVIEWER"},{"updated":"2018-11-19 13:43:11.000000000","updated_by":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"reviewer":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"state":"REVIEWER"},{"updated":"2018-12-10 12:07:00.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2018-12-10 12:18:17.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2019-06-26 09:59:13.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2019-06-26 15:31:20.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-06-27 17:09:40.000000000","updated_by":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"reviewer":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"state":"REVIEWER"},{"updated":"2019-07-01 10:39:01.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2019-07-09 15:50:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"476f4488bf73bb117bd30ead83adccf1fc7ceba8","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-11-19 13:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e97c7bf4dff9a925df2a5bd06543902ef09efeea","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2018-11-19 14:01:02.000000000","message":"Patch Set 1:\n\nThanks for the update!\nIt seems we have lots of duplication here. Can we reduce it?","accounts_in_message":[],"_revision_number":1},{"id":"fb8604bdde426b6bb7852dde57e6c518548f77ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-19 15:13:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/1/check/openstack-tox-docs/536e1bc/html/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/1/check/openstack-tox-pep8/286dfcd/ : FAILURE in 4m 03s\n- kuryr-kubernetes-tempest-octavia http://logs.openstack.org/51/618751/1/check/kuryr-kubernetes-tempest-octavia/5a9a0fd/ : SUCCESS in 1h 00m 59s\n- kuryr-kubernetes-tempest-daemon-octavia http://logs.openstack.org/51/618751/1/check/kuryr-kubernetes-tempest-daemon-octavia/667fb90/ : SUCCESS in 1h 00m 00s\n- kuryr-kubernetes-tempest-daemon-openshift-octavia http://logs.openstack.org/51/618751/1/check/kuryr-kubernetes-tempest-daemon-openshift-octavia/2b2331a/ : FAILURE in 1h 28m 35s (non-voting)\n- kuryr-kubernetes-tempest-daemon-lbaasv2 http://logs.openstack.org/51/618751/1/check/kuryr-kubernetes-tempest-daemon-lbaasv2/59b76df/ : SUCCESS in 43m 38s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia http://logs.openstack.org/51/618751/1/check/kuryr-kubernetes-tempest-daemon-containerized-octavia/9531971/ : SUCCESS in 1h 10m 21s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-ovn http://logs.openstack.org/51/618751/1/check/kuryr-kubernetes-tempest-daemon-containerized-ovn/b54d73b/ : SUCCESS in 1h 07m 36s (non-voting)\n- kuryr-kubernetes-tempest-daemon-octavia-py36 http://logs.openstack.org/51/618751/1/check/kuryr-kubernetes-tempest-daemon-octavia-py36/fa08711/ : SUCCESS in 1h 04m 48s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-py36 http://logs.openstack.org/51/618751/1/check/kuryr-kubernetes-tempest-daemon-containerized-octavia-py36/a01bbd3/ : SUCCESS in 1h 10m 43s\n- kuryr-kubernetes-tempest-multinode-octavia http://logs.openstack.org/51/618751/1/check/kuryr-kubernetes-tempest-multinode-octavia/546a0e9/ : SUCCESS in 1h 14m 50s (non-voting)\n- kuryr-kubernetes-tempest-octavia-centos-7 http://logs.openstack.org/51/618751/1/check/kuryr-kubernetes-tempest-octavia-centos-7/b3ceb55/ : SUCCESS in 1h 03m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b58ee05aeb8756bec9bbe784f4c13be21a5edcd","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-11-20 09:58:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92cc78c0f71f4485d242e4069cf018de4380bb7a","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-11-20 09:59:33.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"a2d935477dd1c58b5852408289f5f99ccc18a00e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-20 12:06:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/2/check/openstack-tox-docs/a971463/html/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/2/check/openstack-tox-pep8/a7dcece/ : SUCCESS in 5m 05s\n- kuryr-kubernetes-tempest-octavia http://logs.openstack.org/51/618751/2/check/kuryr-kubernetes-tempest-octavia/4b29155/ : SUCCESS in 1h 02m 03s\n- kuryr-kubernetes-tempest-daemon-octavia http://logs.openstack.org/51/618751/2/check/kuryr-kubernetes-tempest-daemon-octavia/92144d8/ : SUCCESS in 54m 19s\n- kuryr-kubernetes-tempest-daemon-openshift-octavia http://logs.openstack.org/51/618751/2/check/kuryr-kubernetes-tempest-daemon-openshift-octavia/eee0b9b/ : FAILURE in 1h 20m 50s (non-voting)\n- kuryr-kubernetes-tempest-daemon-lbaasv2 http://logs.openstack.org/51/618751/2/check/kuryr-kubernetes-tempest-daemon-lbaasv2/2298668/ : SUCCESS in 39m 20s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia http://logs.openstack.org/51/618751/2/check/kuryr-kubernetes-tempest-daemon-containerized-octavia/2ebc060/ : SUCCESS in 1h 30m 55s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-ovn http://logs.openstack.org/51/618751/2/check/kuryr-kubernetes-tempest-daemon-containerized-ovn/d6575ff/ : SUCCESS in 1h 19m 03s (non-voting)\n- kuryr-kubernetes-tempest-daemon-octavia-py36 http://logs.openstack.org/51/618751/2/check/kuryr-kubernetes-tempest-daemon-octavia-py36/d4caa73/ : SUCCESS in 1h 22m 36s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-py36 http://logs.openstack.org/51/618751/2/check/kuryr-kubernetes-tempest-daemon-containerized-octavia-py36/fa9982a/ : FAILURE in 1h 57m 43s\n- kuryr-kubernetes-tempest-multinode-octavia http://logs.openstack.org/51/618751/2/check/kuryr-kubernetes-tempest-multinode-octavia/32582d0/ : SUCCESS in 53m 29s (non-voting)\n- kuryr-kubernetes-tempest-octavia-centos-7 http://logs.openstack.org/51/618751/2/check/kuryr-kubernetes-tempest-octavia-centos-7/2cfa4e3/ : SUCCESS in 1h 18m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f712316e70045a3c7de2275f4b42460a55403dd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-20 12:17:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-dragonflow http://logs.openstack.org/51/618751/2/experimental/kuryr-kubernetes-tempest-dragonflow/31a7c8f/ : FAILURE in 38m 54s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia http://logs.openstack.org/51/618751/2/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia/f2438f0/ : FAILURE in 38m 52s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-l2 http://logs.openstack.org/51/618751/2/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-l2/fbb002e/ : SUCCESS in 59m 41s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace http://logs.openstack.org/51/618751/2/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace/7380c33/ : TIMED_OUT in 2h 09m 18s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial http://logs.openstack.org/51/618751/2/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial/1558b9b/ : TIMED_OUT in 2h 04m 25s (non-voting)\n- kuryr-kubernetes-tempest-daemon-ovn http://logs.openstack.org/51/618751/2/experimental/kuryr-kubernetes-tempest-daemon-ovn/8600879/ : FAILURE in 1h 12m 31s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress http://logs.openstack.org/51/618751/2/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress/8734e8d/ : FAILURE in 1h 36m 36s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif http://logs.openstack.org/51/618751/2/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif/8297481/ : FAILURE in 1h 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9fe102d0a51606f0add7b92266353f96166e3f38","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-11-30 11:19:51.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fa877b6fae7d60707ff8bd517a62b0ad70e38955","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-11-30 11:20:00.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"88db023496f1813ab38f85b3747365db1b32aa88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-30 14:36:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/3/check/openstack-tox-docs/91c21a3/html/ : SUCCESS in 5m 45s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/3/check/openstack-tox-pep8/496a862/ : SUCCESS in 6m 43s\n- kuryr-kubernetes-tempest-octavia http://logs.openstack.org/51/618751/3/check/kuryr-kubernetes-tempest-octavia/d0d68ba/ : SUCCESS in 1h 04m 22s\n- kuryr-kubernetes-tempest-daemon-octavia http://logs.openstack.org/51/618751/3/check/kuryr-kubernetes-tempest-daemon-octavia/be9ce71/ : SUCCESS in 1h 02m 39s\n- kuryr-kubernetes-tempest-daemon-openshift-octavia http://logs.openstack.org/51/618751/3/check/kuryr-kubernetes-tempest-daemon-openshift-octavia/c00600d/ : FAILURE in 1h 32m 02s (non-voting)\n- kuryr-kubernetes-tempest-daemon-lbaasv2 http://logs.openstack.org/51/618751/3/check/kuryr-kubernetes-tempest-daemon-lbaasv2/4b7abb7/ : SUCCESS in 48m 22s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia http://logs.openstack.org/51/618751/3/check/kuryr-kubernetes-tempest-daemon-containerized-octavia/4219db1/ : FAILURE in 38m 25s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-ovn http://logs.openstack.org/51/618751/3/check/kuryr-kubernetes-tempest-daemon-containerized-ovn/405c237/ : SUCCESS in 1h 14m 10s (non-voting)\n- kuryr-kubernetes-tempest-daemon-octavia-py36 http://logs.openstack.org/51/618751/3/check/kuryr-kubernetes-tempest-daemon-octavia-py36/23d463e/ : TIMED_OUT in 2h 03m 21s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-py36 http://logs.openstack.org/51/618751/3/check/kuryr-kubernetes-tempest-daemon-containerized-octavia-py36/841c36e/ : SUCCESS in 1h 13m 39s\n- kuryr-kubernetes-tempest-multinode-octavia http://logs.openstack.org/51/618751/3/check/kuryr-kubernetes-tempest-multinode-octavia/f8db781/ : SUCCESS in 1h 04m 27s (non-voting)\n- kuryr-kubernetes-tempest-octavia-centos-7 http://logs.openstack.org/51/618751/3/check/kuryr-kubernetes-tempest-octavia-centos-7/6c6b557/ : SUCCESS in 1h 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5f23cff6db9c317cb9b19445fe7843e699d264cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-30 14:39:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-dragonflow http://logs.openstack.org/51/618751/3/experimental/kuryr-kubernetes-tempest-dragonflow/b8cc469/ : FAILURE in 30m 19s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia http://logs.openstack.org/51/618751/3/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia/5331ee3/ : TIMED_OUT in 2h 05m 14s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-l2 http://logs.openstack.org/51/618751/3/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-l2/a35e036/ : SUCCESS in 1h 11m 43s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace http://logs.openstack.org/51/618751/3/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace/0f6b500/ : FAILURE in 1h 34m 53s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial http://logs.openstack.org/51/618751/3/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial/54a7b1d/ : TIMED_OUT in 2h 04m 26s (non-voting)\n- kuryr-kubernetes-tempest-daemon-ovn http://logs.openstack.org/51/618751/3/experimental/kuryr-kubernetes-tempest-daemon-ovn/eedced4/ : TIMED_OUT in 2h 08m 02s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress http://logs.openstack.org/51/618751/3/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress/d1e1d5d/ : TIMED_OUT in 2h 06m 13s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif http://logs.openstack.org/51/618751/3/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif/1fd0bb9/ : FAILURE in 1h 36m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a58aa0a54b029cddc9bebc2479a911a743c11d8","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-11-30 15:48:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7a9583a787125d48c4b44df2e71afeda1baef372","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-11-30 15:49:03.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"f598a0a049297d913daac93e8aec0f4c59279451","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-30 18:47:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-dragonflow http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-dragonflow/b59c577/ : FAILURE in 29m 35s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia/508a5c5/ : POST_FAILURE in 1h 21m 40s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-l2 http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-l2/2ddb63a/ : SUCCESS in 1h 07m 08s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace/978eced/ : FAILURE in 2h 00m 21s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial/8b4d8f6/ : SUCCESS in 2h 00m 42s (non-voting)\n- kuryr-kubernetes-tempest-daemon-ovn http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-ovn/369b207/ : SUCCESS in 1h 10m 56s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress/484bc9f/ : FAILURE in 1h 44m 32s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif/e19808f/ : FAILURE in 1h 42m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4db571259c04c8546c5f6b0f444f61c80e9f8b16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-30 18:51:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/4/check/openstack-tox-docs/24d78a5/html/ : SUCCESS in 5m 06s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/4/check/openstack-tox-pep8/b90a1c2/ : SUCCESS in 5m 06s\n- kuryr-kubernetes-tempest-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-octavia/aab0c5e/ : SUCCESS in 53m 42s\n- kuryr-kubernetes-tempest-daemon-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-octavia/2069e6f/ : SUCCESS in 1h 19m 10s\n- kuryr-kubernetes-tempest-daemon-openshift-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-openshift-octavia/881a1bd/ : SUCCESS in 1h 27m 51s (non-voting)\n- kuryr-kubernetes-tempest-daemon-lbaasv2 http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-lbaasv2/30706f5/ : SUCCESS in 47m 20s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-containerized-octavia/d3faf9b/ : SUCCESS in 1h 13m 00s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-ovn http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-containerized-ovn/ee98994/ : SUCCESS in 1h 21m 09s (non-voting)\n- kuryr-kubernetes-tempest-daemon-octavia-py36 http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-octavia-py36/02c8b25/ : SUCCESS in 1h 08m 05s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-py36 http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-containerized-octavia-py36/7be9bb2/ : FAILURE in 1h 37m 55s\n- kuryr-kubernetes-tempest-multinode-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-multinode-octavia/f01c4a3/ : FAILURE in 1h 19m 04s (non-voting)\n- kuryr-kubernetes-tempest-octavia-centos-7 http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-octavia-centos-7/4316531/ : SUCCESS in 1h 18m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9ef075cdd0c637c6149e4a476e1c9ed9e50f08c4","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-12-03 08:16:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c18b84495822a902e1c1bd6d8318e3dbc02b69fa","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-12-03 08:16:07.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"c09edbc2dcf2278ea2183c4993ffe4de09e579b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-03 10:25:50.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/4/check/openstack-tox-docs/67a8806/html/ : SUCCESS in 6m 04s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/4/check/openstack-tox-pep8/a30e935/ : SUCCESS in 5m 04s\n- kuryr-kubernetes-tempest-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-octavia/7a2212c/ : SUCCESS in 56m 37s\n- kuryr-kubernetes-tempest-daemon-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-octavia/eb4735c/ : SUCCESS in 1h 04m 15s\n- kuryr-kubernetes-tempest-daemon-openshift-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-openshift-octavia/ec55774/ : TIMED_OUT in 2h 05m 26s (non-voting)\n- kuryr-kubernetes-tempest-daemon-lbaasv2 http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-lbaasv2/4074acc/ : SUCCESS in 53m 42s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-containerized-octavia/1b6d8c5/ : SUCCESS in 1h 09m 40s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-ovn http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-containerized-ovn/f864942/ : TIMED_OUT in 2h 03m 14s (non-voting)\n- kuryr-kubernetes-tempest-daemon-octavia-py36 http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-octavia-py36/fd8e770/ : SUCCESS in 53m 28s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-py36 http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-containerized-octavia-py36/03f18d8/ : TIMED_OUT in 2h 03m 39s\n- kuryr-kubernetes-tempest-multinode-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-multinode-octavia/2967d8c/ : SUCCESS in 1h 09m 33s (non-voting)\n- kuryr-kubernetes-tempest-octavia-centos-7 http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-octavia-centos-7/4bdbf83/ : SUCCESS in 1h 08m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"41cbbd5ead8f0df45b6eb8e6a374338e394699d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-03 10:28:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-dragonflow http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-dragonflow/d697010/ : FAILURE in 42m 05s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia/9da0578/ : FAILURE in 1h 23m 51s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-l2 http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-l2/8eefa39/ : SUCCESS in 1h 23m 40s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace/d1f214a/ : FAILURE in 42m 50s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial/0045be8/ : SUCCESS in 1h 49m 23s (non-voting)\n- kuryr-kubernetes-tempest-daemon-ovn http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-ovn/ff00ef1/ : SUCCESS in 1h 01m 57s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress/7941c21/ : TIMED_OUT in 2h 05m 07s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif/0cb1420/ : FAILURE in 1h 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d7ae7d08566db1d16fc98e2aecbf5cf7ea58c5fb","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-12-03 10:36:15.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9e4fd23e542bd34e41dbf67bf9bde4aab253a9ab","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-12-03 10:36:23.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"46b4959dcb2fb1f0f5d53656b213a58f0c77b6f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-03 13:47:51.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/4/check/openstack-tox-docs/fdd93da/html/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/4/check/openstack-tox-pep8/0c58c02/ : SUCCESS in 4m 28s\n- kuryr-kubernetes-tempest-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-octavia/0247e83/ : FAILURE in 42m 10s\n- kuryr-kubernetes-tempest-daemon-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-octavia/23762cf/ : SUCCESS in 1h 12m 30s\n- kuryr-kubernetes-tempest-daemon-openshift-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-openshift-octavia/2d8eb68/ : FAILURE in 1h 33m 44s (non-voting)\n- kuryr-kubernetes-tempest-daemon-lbaasv2 http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-lbaasv2/d58442a/ : SUCCESS in 32m 16s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-containerized-octavia/877d177/ : SUCCESS in 1h 22m 03s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-ovn http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-containerized-ovn/f120572/ : SUCCESS in 1h 44m 54s (non-voting)\n- kuryr-kubernetes-tempest-daemon-octavia-py36 http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-octavia-py36/a3b8865/ : SUCCESS in 1h 22m 32s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-py36 http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-daemon-containerized-octavia-py36/43a735e/ : SUCCESS in 1h 13m 00s\n- kuryr-kubernetes-tempest-multinode-octavia http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-multinode-octavia/709bcd3/ : SUCCESS in 1h 13m 22s (non-voting)\n- kuryr-kubernetes-tempest-octavia-centos-7 http://logs.openstack.org/51/618751/4/check/kuryr-kubernetes-tempest-octavia-centos-7/bef9c7f/ : SUCCESS in 1h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8223445775c384450597e20ddbf9303bb130ae01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-03 14:03:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-dragonflow http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-dragonflow/8f70828/ : FAILURE in 28m 32s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia/cf4a3f6/ : FAILURE in 41m 34s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-l2 http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-l2/407821c/ : SUCCESS in 1h 17m 30s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace/54fe6de/ : FAILURE in 1h 31m 38s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial/316ebc2/ : FAILURE in 2h 01m 22s (non-voting)\n- kuryr-kubernetes-tempest-daemon-ovn http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-ovn/b728643/ : SUCCESS in 1h 17m 12s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress/8c67735/ : FAILURE in 1h 53m 06s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif http://logs.openstack.org/51/618751/4/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif/0406661/ : SUCCESS in 1h 20m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6965f77c743f1e13ee1f3a95d6f928102da499f9","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-12-07 08:56:28.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"83f643608663019a1328909184a9742065fa2a50","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-12-07 08:56:35.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"8a5a151daa284d586ef5eb6dd1071afa56b26a5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-07 11:11:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/5/check/openstack-tox-docs/b480837/html/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/5/check/openstack-tox-pep8/31e8806/ : SUCCESS in 7m 02s\n- kuryr-kubernetes-tempest-octavia http://logs.openstack.org/51/618751/5/check/kuryr-kubernetes-tempest-octavia/2f5bb3f/ : FAILURE in 1h 06m 52s\n- kuryr-kubernetes-tempest-daemon-octavia http://logs.openstack.org/51/618751/5/check/kuryr-kubernetes-tempest-daemon-octavia/dc63e6d/ : SUCCESS in 1h 13m 06s\n- kuryr-kubernetes-tempest-daemon-openshift-octavia http://logs.openstack.org/51/618751/5/check/kuryr-kubernetes-tempest-daemon-openshift-octavia/445be0a/ : SUCCESS in 1h 18m 03s (non-voting)\n- kuryr-kubernetes-tempest-daemon-lbaasv2 http://logs.openstack.org/51/618751/5/check/kuryr-kubernetes-tempest-daemon-lbaasv2/6b2b804/ : SUCCESS in 55m 07s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia http://logs.openstack.org/51/618751/5/check/kuryr-kubernetes-tempest-daemon-containerized-octavia/1a308ab/ : SUCCESS in 1h 19m 23s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-ovn http://logs.openstack.org/51/618751/5/check/kuryr-kubernetes-tempest-daemon-containerized-ovn/f2044f6/ : SUCCESS in 1h 16m 45s (non-voting)\n- kuryr-kubernetes-tempest-daemon-octavia-py36 http://logs.openstack.org/51/618751/5/check/kuryr-kubernetes-tempest-daemon-octavia-py36/d476330/ : TIMED_OUT in 2h 04m 01s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-py36 http://logs.openstack.org/51/618751/5/check/kuryr-kubernetes-tempest-daemon-containerized-octavia-py36/732599e/ : SUCCESS in 1h 11m 26s\n- kuryr-kubernetes-tempest-multinode-octavia http://logs.openstack.org/51/618751/5/check/kuryr-kubernetes-tempest-multinode-octavia/2a6724d/ : SUCCESS in 1h 10m 58s (non-voting)\n- kuryr-kubernetes-tempest-octavia-centos-7 http://logs.openstack.org/51/618751/5/check/kuryr-kubernetes-tempest-octavia-centos-7/1c85eff/ : SUCCESS in 1h 06m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"49bebc42efb841184b0558f269e1947bb94ee4f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-07 11:16:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-dragonflow http://logs.openstack.org/51/618751/5/experimental/kuryr-kubernetes-tempest-dragonflow/fbcb764/ : FAILURE in 31m 54s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia http://logs.openstack.org/51/618751/5/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia/c41ace6/ : FAILURE in 1h 44m 15s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-l2 http://logs.openstack.org/51/618751/5/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-l2/3919075/ : SUCCESS in 1h 04m 11s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace http://logs.openstack.org/51/618751/5/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace/80f4bce/ : TIMED_OUT in 2h 04m 24s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-network-policy http://logs.openstack.org/51/618751/5/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-network-policy/3851f2d/ : SUCCESS in 46m 41s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial http://logs.openstack.org/51/618751/5/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial/4d7a3a7/ : SUCCESS in 1h 57m 19s (non-voting)\n- kuryr-kubernetes-tempest-daemon-ovn http://logs.openstack.org/51/618751/5/experimental/kuryr-kubernetes-tempest-daemon-ovn/40f2d22/ : SUCCESS in 1h 06m 24s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress http://logs.openstack.org/51/618751/5/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress/f75e422/ : FAILURE in 1h 54m 00s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif http://logs.openstack.org/51/618751/5/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif/66f9aa1/ : FAILURE in 1h 26m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bc6de202be320346fa5cc1d71eea5cfedb39b02d","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-12-10 08:03:54.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"32c20fa41f4ddef81dcd456605f1611a5b2a84a8","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-12-10 08:04:00.000000000","message":"Patch Set 6:\n\ncheck experimental","accounts_in_message":[],"_revision_number":6},{"id":"d72fa1d21f603f724bfab2d6f6a8d83d1e4dee7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-10 09:47:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/6/check/openstack-tox-docs/15eb163/html/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/6/check/openstack-tox-pep8/228d0d3/ : SUCCESS in 5m 13s\n- kuryr-kubernetes-tempest-octavia http://logs.openstack.org/51/618751/6/check/kuryr-kubernetes-tempest-octavia/1b0110e/ : SUCCESS in 59m 44s\n- kuryr-kubernetes-tempest-daemon-octavia http://logs.openstack.org/51/618751/6/check/kuryr-kubernetes-tempest-daemon-octavia/2d46b21/ : SUCCESS in 54m 58s\n- kuryr-kubernetes-tempest-daemon-openshift-octavia http://logs.openstack.org/51/618751/6/check/kuryr-kubernetes-tempest-daemon-openshift-octavia/b41965e/ : FAILURE in 1h 37m 28s (non-voting)\n- kuryr-kubernetes-tempest-daemon-lbaasv2 http://logs.openstack.org/51/618751/6/check/kuryr-kubernetes-tempest-daemon-lbaasv2/19818a8/ : SUCCESS in 37m 28s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia http://logs.openstack.org/51/618751/6/check/kuryr-kubernetes-tempest-daemon-containerized-octavia/34286bd/ : SUCCESS in 1h 06m 47s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-ovn http://logs.openstack.org/51/618751/6/check/kuryr-kubernetes-tempest-daemon-containerized-ovn/de3b1a0/ : SUCCESS in 1h 03m 35s (non-voting)\n- kuryr-kubernetes-tempest-daemon-octavia-py36 http://logs.openstack.org/51/618751/6/check/kuryr-kubernetes-tempest-daemon-octavia-py36/67cd55b/ : SUCCESS in 58m 31s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-py36 http://logs.openstack.org/51/618751/6/check/kuryr-kubernetes-tempest-daemon-containerized-octavia-py36/33ad30b/ : SUCCESS in 1h 22m 34s\n- kuryr-kubernetes-tempest-multinode-octavia http://logs.openstack.org/51/618751/6/check/kuryr-kubernetes-tempest-multinode-octavia/c4c3e1b/ : SUCCESS in 59m 27s (non-voting)\n- kuryr-kubernetes-tempest-octavia-centos-7 http://logs.openstack.org/51/618751/6/check/kuryr-kubernetes-tempest-octavia-centos-7/1fc01f5/ : SUCCESS in 1h 08m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f7e1ab8b7647bffbbba03b194cbbc6224aa5f7b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-10 10:11:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-dragonflow http://logs.openstack.org/51/618751/6/experimental/kuryr-kubernetes-tempest-dragonflow/6ca9f84/ : FAILURE in 26m 48s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia http://logs.openstack.org/51/618751/6/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia/ee4b549/ : FAILURE in 33m 22s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-l2 http://logs.openstack.org/51/618751/6/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-l2/6d03e8f/ : SUCCESS in 1h 16m 46s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace http://logs.openstack.org/51/618751/6/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace/6ded73a/ : FAILURE in 1h 38m 29s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-network-policy http://logs.openstack.org/51/618751/6/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-network-policy/bd91592/ : FAILURE in 58m 56s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial http://logs.openstack.org/51/618751/6/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial/25b8642/ : FAILURE in 1h 57m 37s (non-voting)\n- kuryr-kubernetes-tempest-daemon-ovn http://logs.openstack.org/51/618751/6/experimental/kuryr-kubernetes-tempest-daemon-ovn/2e42ab3/ : SUCCESS in 1h 21m 32s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress http://logs.openstack.org/51/618751/6/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress/5c2f841/ : TIMED_OUT in 2h 05m 05s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif http://logs.openstack.org/51/618751/6/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif/d2883f2/ : FAILURE in 1h 20m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3276bba510bee369474a40b394d666290a499e65","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2018-12-10 10:34:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ca68aabfef8e991370134abf88d702560cf60d97","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2018-12-10 10:34:14.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0e166a226fdbd8f3ee5aafef6210787c6c1f0fc7","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-12-10 10:57:52.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ed213772da9e609c8fb2281d53058a80c2a5ac64","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2018-12-10 12:07:00.000000000","message":"Patch Set 6: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6d0d054d3499f09bd67708583a7a21e3fc2d0c9d","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2018-12-10 12:18:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c9548d0e4f81f5350118d42e9784030440343efb","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-12-24 13:35:48.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"1fb15894dd8a26914722d5aca4b0e6af407681d3","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-12-24 13:36:00.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"0ab7d0638543e9592569c6a31145891c9b6dc9d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-24 15:17:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/7/check/openstack-tox-docs/0439d84/html/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/7/check/openstack-tox-pep8/ac3b7db/ : SUCCESS in 3m 49s\n- kuryr-kubernetes-tempest-octavia http://logs.openstack.org/51/618751/7/check/kuryr-kubernetes-tempest-octavia/4b3ace3/ : SUCCESS in 1h 02m 08s\n- kuryr-kubernetes-tempest-daemon-octavia http://logs.openstack.org/51/618751/7/check/kuryr-kubernetes-tempest-daemon-octavia/c818de2/ : SUCCESS in 58m 44s\n- kuryr-kubernetes-tempest-daemon-openshift-octavia http://logs.openstack.org/51/618751/7/check/kuryr-kubernetes-tempest-daemon-openshift-octavia/9badfab/ : FAILURE in 1h 35m 53s (non-voting)\n- kuryr-kubernetes-tempest-daemon-lbaasv2 http://logs.openstack.org/51/618751/7/check/kuryr-kubernetes-tempest-daemon-lbaasv2/aab642a/ : SUCCESS in 42m 45s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia http://logs.openstack.org/51/618751/7/check/kuryr-kubernetes-tempest-daemon-containerized-octavia/9aba2c2/ : SUCCESS in 1h 05m 29s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-ovn http://logs.openstack.org/51/618751/7/check/kuryr-kubernetes-tempest-daemon-containerized-ovn/132a7c3/ : SUCCESS in 1h 19m 04s (non-voting)\n- kuryr-kubernetes-tempest-daemon-octavia-py36 http://logs.openstack.org/51/618751/7/check/kuryr-kubernetes-tempest-daemon-octavia-py36/0aec44d/ : SUCCESS in 1h 02m 47s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-py36 http://logs.openstack.org/51/618751/7/check/kuryr-kubernetes-tempest-daemon-containerized-octavia-py36/4533b97/ : SUCCESS in 1h 09m 00s\n- kuryr-kubernetes-tempest-multinode-daemon-octavia-containerized http://logs.openstack.org/51/618751/7/check/kuryr-kubernetes-tempest-multinode-daemon-octavia-containerized/fc6af85/ : SUCCESS in 1h 17m 06s (non-voting)\n- kuryr-kubernetes-tempest-octavia-centos-7 http://logs.openstack.org/51/618751/7/check/kuryr-kubernetes-tempest-octavia-centos-7/dbbfb40/ : SUCCESS in 1h 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"04313d0e4d0b92d8b4bd285f7eae6225e2ecf49b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-24 15:35:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-dragonflow http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-dragonflow/033d807/ : FAILURE in 26m 52s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia/9aafada/ : SUCCESS in 1h 14m 16s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-l2 http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-l2/66637df/ : SUCCESS in 1h 09m 39s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace/131ee90/ : FAILURE in 1h 46m 37s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-network-policy http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-network-policy/d27d5e7/ : SUCCESS in 45m 14s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial/dd3e3a1/ : FAILURE in 1h 57m 48s (non-voting)\n- kuryr-kubernetes-tempest-daemon-ovn http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-ovn/7bad375/ : SUCCESS in 1h 06m 31s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress/476790c/ : FAILURE in 1h 44m 00s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif/f3166d3/ : FAILURE in 1h 11m 21s (non-voting)\n- kuryr-kubernetes-tempest-multinode-daemon-octavia-ha http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-multinode-daemon-octavia-ha/19233a0/ : SUCCESS in 1h 55m 44s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-crio http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-crio/f39b9fe/ : SUCCESS in 1h 04m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"15f09e883ee14a8a9bbd384f666c9c05c778abc2","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2018-12-24 15:52:31.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"898d1ae8ee9c0af78dd4f9bc230d31630b56dd10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-24 18:00:47.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-dragonflow http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-dragonflow/3ce65df/ : FAILURE in 23m 40s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia/e24a95f/ : FAILURE in 1h 22m 19s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-l2 http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-l2/d2d807e/ : SUCCESS in 1h 05m 37s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-pools-namespace/7d9c231/ : FAILURE in 1h 27m 54s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-network-policy http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-network-policy/03ac6c1/ : SUCCESS in 40m 08s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-containerized-openshift-octavia-serial/59ee163/ : FAILURE in 1h 54m 36s (non-voting)\n- kuryr-kubernetes-tempest-daemon-ovn http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-ovn/f3ecbb1/ : FAILURE in 1h 18m 55s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-ingress/8c20f89/ : FAILURE in 2h 03m 14s (non-voting)\n- kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-openshift-octavia-multi-vif/9cb5d34/ : FAILURE in 1h 30m 01s (non-voting)\n- kuryr-kubernetes-tempest-multinode-daemon-octavia-ha http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-multinode-daemon-octavia-ha/b76e02c/ : SUCCESS in 1h 44m 53s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-crio http://logs.openstack.org/51/618751/7/experimental/kuryr-kubernetes-tempest-daemon-containerized-octavia-crio/222bcf2/ : SUCCESS in 58m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d22b3dbe83a11b6176df5822351f00442455c9ad","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-05-24 10:13:28.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ec6b792b145d37dae2c5fbd8f3aae6dd6c475b52","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-05-24 10:14:53.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"502b64acefa574f3f0687b75f9226930572dbc12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 12:06:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/8/check/openstack-tox-docs/b9e3224/html/ : SUCCESS in 4m 26s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/8/check/openstack-tox-pep8/25d69ce/ : SUCCESS in 4m 20s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/8/check/kuryr-kubernetes-tempest/ca93dbf/ : SUCCESS in 1h 20m 42s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/8/check/kuryr-kubernetes-tempest-openshift/f032508/ : FAILURE in 1h 36m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/8/check/kuryr-kubernetes-tempest-containerized/da12d6f/ : SUCCESS in 1h 46m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/8/check/kuryr-kubernetes-tempest-containerized-ovn/d39ad19/ : FAILURE in 1h 49m 54s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/8/check/kuryr-kubernetes-tempest-py36/63d437d/ : SUCCESS in 1h 25m 36s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/8/check/kuryr-kubernetes-tempest-containerized-py36/f218ca9/ : SUCCESS in 1h 51m 22s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/8/check/kuryr-kubernetes-tempest-multinode-containerized/70ba4a1/ : SUCCESS in 1h 41m 20s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/8/check/kuryr-kubernetes-tempest-centos-7/61c0b2e/ : FAILURE in 22m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"79006fbd836cc400b2cf0c82653419a51b22eee5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 13:25:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-containerized-openshift/0ba080c/ : TIMED_OUT in 3h 05m 44s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-containerized-l2/e3c9b1e/ : FAILURE in 2h 11m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/065d981/ : TIMED_OUT in 3h 07m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-containerized-network-policy/791d63d/ : SUCCESS in 1h 00m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/97b50a9/ : FAILURE in 2h 48m 14s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-ovn/9e3cd87/ : FAILURE in 1h 29m 54s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-openshift-ingress/92a507a/ : FAILURE in 57m 58s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/a7352db/ : FAILURE in 2h 24m 56s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-multinode-ha/b8467dd/ : TIMED_OUT in 2h 14m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-containerized-crio/1de22f9/ : FAILURE in 26m 03s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"923f7743f1afbc0e56bee6116b59627a2c6920ac","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-05-24 14:52:37.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"e1b83353821362e35a4bca010fdf8c2bbdf99802","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 18:03:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-containerized-openshift/9e217da/ : FAILURE in 1h 50m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-containerized-l2/f27da95/ : FAILURE in 1h 56m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/7227121/ : TIMED_OUT in 3h 04m 01s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-containerized-network-policy/1bffb5c/ : SUCCESS in 55m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/cbf05e0/ : TIMED_OUT in 3h 04m 32s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-ovn/ddcada3/ : FAILURE in 1h 40m 41s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-openshift-ingress/43f8b40/ : FAILURE in 59m 51s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/9f184f7/ : FAILURE in 1h 44m 23s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-multinode-ha/287b21f/ : TIMED_OUT in 2h 15m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/8/experimental/kuryr-kubernetes-tempest-containerized-crio/09840cf/ : FAILURE in 25m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1509541158d7fc9ed7c4fce4427b403c8cdbe0e8","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-04 06:34:43.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"af9652457e89844ea3cc36389a0b022477cdea29","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-04 06:34:47.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"bb240d8ad3ce2e72f59a90c687384e0510f28966","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-04 06:34:53.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"8d2cf919c447e4ad75df9e25805a9d1b76372e13","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-04 06:34:58.000000000","message":"Patch Set 9:\n\ncheck experimental","accounts_in_message":[],"_revision_number":9},{"id":"3aecaaba94837ebceb5adcc49fba62795ed76ae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 09:04:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/9/check/openstack-tox-docs/1ce8789/html/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/9/check/openstack-tox-pep8/f73baae/ : SUCCESS in 4m 20s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/9/check/kuryr-kubernetes-tempest/9c069e7/ : SUCCESS in 1h 41m 17s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/9/check/kuryr-kubernetes-tempest-openshift/2143e4d/ : FAILURE in 2h 24m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/9/check/kuryr-kubernetes-tempest-containerized/fa30011/ : SUCCESS in 1h 36m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/9/check/kuryr-kubernetes-tempest-containerized-ovn/55f94d2/ : FAILURE in 1h 38m 21s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/9/check/kuryr-kubernetes-tempest-py36/3b4a94c/ : SUCCESS in 1h 38m 37s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/9/check/kuryr-kubernetes-tempest-containerized-py36/a9dea09/ : SUCCESS in 1h 40m 16s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/9/check/kuryr-kubernetes-tempest-multinode-containerized/63aeee5/ : SUCCESS in 1h 50m 32s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/9/check/kuryr-kubernetes-tempest-centos-7/282922a/ : SUCCESS in 1h 35m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f52a8b82aa4be04cdabba663ceaff988015ebc18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 09:59:24.000000000","message":"Patch Set 9:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/9/experimental/kuryr-kubernetes-tempest-containerized-openshift/873b343/ : FAILURE in 1h 42m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/9/experimental/kuryr-kubernetes-tempest-containerized-l2/09a60d7/ : FAILURE in 1h 45m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/9/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/74502f5/ : TIMED_OUT in 3h 03m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/9/experimental/kuryr-kubernetes-tempest-containerized-network-policy/5a40159/ : SUCCESS in 59m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/9/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/1ca6170/ : TIMED_OUT in 3h 08m 19s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/9/experimental/kuryr-kubernetes-tempest-ovn/bea4561/ : FAILURE in 1h 53m 33s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/9/experimental/kuryr-kubernetes-tempest-openshift-ingress/1bcc532/ : FAILURE in 1h 11m 46s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/9/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/0757471/ : FAILURE in 1h 59m 16s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/9/experimental/kuryr-kubernetes-tempest-multinode-ha/7f9b143/ : TIMED_OUT in 2h 15m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/9/experimental/kuryr-kubernetes-tempest-containerized-crio/faae55b/ : FAILURE in 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"708aeba06eb62d1d6fe2f923d0231af3db908cb9","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-11 07:07:48.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"00872d73c399013abbc88be8d4780b4d14d584f1","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-11 07:09:25.000000000","message":"Patch Set 10:\n\ncheck experimental","accounts_in_message":[],"_revision_number":10},{"id":"16cf228427e8a5b2221d243ce2d00b4ecf894daf","author":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"date":"2019-06-11 09:48:01.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"aedec8512d021fa46af51c94c42b7f2da95f2917","author":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"date":"2019-06-11 09:48:05.000000000","message":"Patch Set 10: Code-Review-1","accounts_in_message":[],"_revision_number":10},{"id":"ef579c99c1cdd751a848028ec1a72deebd5ff87b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 10:24:40.000000000","message":"Patch Set 10:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/10/experimental/kuryr-kubernetes-tempest-containerized-openshift/187b984/ : TIMED_OUT in 3h 04m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/10/experimental/kuryr-kubernetes-tempest-containerized-l2/8e1c4a5/ : FAILURE in 1h 45m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/10/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/0de6168/ : TIMED_OUT in 3h 04m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/10/experimental/kuryr-kubernetes-tempest-containerized-network-policy/e793829/ : SUCCESS in 59m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/10/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/945bdaa/ : FAILURE in 1h 49m 28s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/10/experimental/kuryr-kubernetes-tempest-ovn/851e721/ : FAILURE in 1h 43m 10s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/10/experimental/kuryr-kubernetes-tempest-openshift-ingress/62e33b5/ : FAILURE in 1h 03m 13s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/10/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/962d8b6/ : FAILURE in 1h 31m 16s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/10/experimental/kuryr-kubernetes-tempest-multinode-ha/92bf618/ : TIMED_OUT in 2h 16m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/10/experimental/kuryr-kubernetes-tempest-containerized-crio/c29de98/ : FAILURE in 26m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"46476cf944ebc7aa6ffbd12c280863c3d0625fed","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-11 10:25:56.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"4270988dad47032780571b83eb1b9f899928a553","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-11 10:26:00.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8837a57ad23c90af2705abf22d0c9577097b7bdb","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-11 10:26:11.000000000","message":"Patch Set 11:\n\ncheck experimental","accounts_in_message":[],"_revision_number":11},{"id":"05192cae9d4d1698b4bcc8e3fd7c94cfb488db52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 12:22:49.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/11/check/openstack-tox-docs/ad7dc1a/html/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/11/check/openstack-tox-pep8/7ad2c54/ : SUCCESS in 5m 15s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest/4c0a5ce/ : SUCCESS in 1h 16m 50s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-openshift/5e0bfa9/ : FAILURE in 1h 34m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-containerized/04ea11d/ : SUCCESS in 1h 46m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-containerized-ovn/f5954a5/ : FAILURE in 1h 46m 51s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-py36/8c7543e/ : SUCCESS in 1h 18m 27s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-containerized-py36/4df4ea8/ : FAILURE in 1h 45m 33s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-multinode-containerized/78022c6/ : SUCCESS in 1h 48m 52s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-centos-7/2005f57/ : SUCCESS in 1h 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"36f030c74f24c08f52a57bbc692d5213ec4decfc","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-11 12:26:21.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"59a32cdedf007fcb6cd585ba901e6fd73bb63e4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 13:36:08.000000000","message":"Patch Set 11:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/11/experimental/kuryr-kubernetes-tempest-containerized-openshift/83edf84/ : FAILURE in 1h 36m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/11/experimental/kuryr-kubernetes-tempest-containerized-l2/b39f223/ : FAILURE in 2h 12m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/11/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/ec382ee/ : TIMED_OUT in 3h 06m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/11/experimental/kuryr-kubernetes-tempest-containerized-network-policy/9255fcb/ : SUCCESS in 1h 02m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/11/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/c538334/ : FAILURE in 2h 08m 02s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/11/experimental/kuryr-kubernetes-tempest-ovn/b70c16c/ : FAILURE in 1h 44m 42s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/11/experimental/kuryr-kubernetes-tempest-openshift-ingress/63d6254/ : FAILURE in 1h 04m 11s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/11/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/7400b09/ : FAILURE in 1h 42m 08s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/11/experimental/kuryr-kubernetes-tempest-multinode-ha/142b8df/ : TIMED_OUT in 2h 16m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/11/experimental/kuryr-kubernetes-tempest-containerized-crio/ea4daf3/ : FAILURE in 32m 47s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c640f5eb89b405e87bb7f172a50e8a15344c6877","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 14:33:44.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/11/check/openstack-tox-docs/57a9f4a/html/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/11/check/openstack-tox-pep8/56ca49a/ : SUCCESS in 5m 27s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest/b9a0bfe/ : SUCCESS in 1h 41m 27s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-openshift/6a2917c/ : FAILURE in 1h 26m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-containerized/286fea7/ : SUCCESS in 1h 55m 54s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-containerized-ovn/1c31c3c/ : FAILURE in 1h 45m 01s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-py36/fb2747e/ : SUCCESS in 1h 19m 04s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-containerized-py36/8fe74e0/ : SUCCESS in 1h 50m 00s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-multinode-containerized/430dafe/ : SUCCESS in 1h 44m 03s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/11/check/kuryr-kubernetes-tempest-centos-7/58215f1/ : SUCCESS in 1h 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ff1bc812ab1418862450a6c3dfbd82bc24ebd875","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-13 07:09:40.000000000","message":"Uploaded patch set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"fff8312bb699e571840ab41a19918e4bfbbacc93","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-13 07:10:41.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"76a3885e88f570b5b018eb934c06c2e54b0ece0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:37:04.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/12/check/openstack-tox-docs/21f01af/html/ : SUCCESS in 4m 46s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/12/check/openstack-tox-pep8/a44c4b9/ : SUCCESS in 4m 23s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest/6ba7b08/ : SUCCESS in 1h 31m 42s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-openshift/88a5a22/ : FAILURE in 1h 24m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized/592624e/ : SUCCESS in 1h 35m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-ovn/3efc564/ : SUCCESS in 1h 47m 48s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-py36/06244e2/ : FAILURE in 2h 01m 24s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-py36/b6b60dd/ : SUCCESS in 1h 33m 21s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-multinode-containerized/2add61d/ : SUCCESS in 1h 58m 33s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-centos-7/18770ea/ : SUCCESS in 1h 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b7f1b2af673a3957520cce7cb81a1c47569faf17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:38:49.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"90204ba5205504d86b15b5308d663536d8cbb853","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-13 09:53:01.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"b5ce956fede10c2197ec24f3d188cdfc2f31aafe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 11:55:38.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/12/check/openstack-tox-docs/8b4acbe/html/ : SUCCESS in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/12/check/openstack-tox-pep8/ceec6e2/ : SUCCESS in 5m 26s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest/3d7fa8b/ : SUCCESS in 1h 06m 02s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-openshift/9d3e94b/ : FAILURE in 1h 45m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized/321295e/ : SUCCESS in 1h 56m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-ovn/f8b0f59/ : SUCCESS in 1h 38m 00s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-py36/7b1b97e/ : FAILURE in 1h 51m 30s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-py36/60c38b2/ : SUCCESS in 1h 42m 54s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-multinode-containerized/521a550/ : SUCCESS in 1h 43m 10s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-centos-7/c3f182e/ : SUCCESS in 1h 36m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ab6bbb586c4c8e748e30588b0f8314c06834113f","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-13 12:56:54.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"d0777264e3b5e62da77f23bb6452dd54e0280e75","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-13 14:15:32.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"9188429f86aee0e02ee3027986e0624c250ecea1","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-13 15:54:29.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"2ec75eb5e13abb6295b5b7c30ae32261702ae336","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 20:49:11.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/12/check/openstack-tox-docs/acb1462/html/ : SUCCESS in 7m 49s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/12/check/openstack-tox-pep8/3e1fd0c/ : SUCCESS in 5m 26s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest/030df8b/ : FAILURE in 1h 38m 21s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-openshift/95fee78/ : FAILURE in 2h 04m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized/f5092da/ : SUCCESS in 1h 38m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-ovn/123fa49/ : TIMED_OUT in 3h 05m 31s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-py36/5be31b0/ : SUCCESS in 1h 30m 45s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-py36/c56c3f1/ : FAILURE in 2h 01m 39s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-multinode-containerized/2f952b6/ : SUCCESS in 1h 53m 57s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-centos-7/4dc0883/ : SUCCESS in 1h 36m 01s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"00f2130ace1f95a72cc4d8ed6da09002fb5cf8f5","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-14 09:21:59.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"d26f2dc8a91af3192239d2d6180b866287787930","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-14 09:22:14.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"d816fad1ff854bbaea35a1798684ee8e8a073b13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 12:34:08.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/12/check/openstack-tox-docs/7b1515e/html/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/12/check/openstack-tox-pep8/db2d0af/ : SUCCESS in 5m 04s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest/8f1dcf7/ : SUCCESS in 1h 36m 34s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-openshift/a847a31/ : FAILURE in 1h 53m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized/8e78f9d/ : SUCCESS in 1h 24m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-ovn/9dc6ea6/ : TIMED_OUT in 3h 04m 43s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-py36/7bfd323/ : FAILURE in 1h 01m 49s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-py36/39b5c2c/ : FAILURE in 1h 43m 09s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-multinode-containerized/6a7238d/ : SUCCESS in 1h 53m 59s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-centos-7/6fd44e2/ : SUCCESS in 1h 46m 31s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b3a79a7c6f26e32f2ee96f8c7be05c46aee4bcd6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 12:36:49.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-openshift/e65ba89/ : FAILURE in 1h 46m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-l2/8fa4a99/ : FAILURE in 2h 10m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/e5bd566/ : TIMED_OUT in 3h 05m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-network-policy/9e2036c/ : SUCCESS in 1h 03m 41s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/d2b3da6/ : FAILURE in 2h 33m 15s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-ovn/4a47fc5/ : SUCCESS in 1h 29m 34s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-openshift-ingress/4c5e0f3/ : FAILURE in 1h 06m 06s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/c0595af/ : FAILURE in 1h 37m 38s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-multinode-ha/f168799/ : TIMED_OUT in 2h 15m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-crio/c9fd1bd/ : FAILURE in 25m 30s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"61d75a4223a57fc0c59045083692b175aba35292","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-14 12:55:31.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"a8680b7277da780921f303f542e9b6166854ccbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 18:57:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/12/check/openstack-tox-docs/d57e689/html/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/12/check/openstack-tox-pep8/9eba70e/ : SUCCESS in 4m 16s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest/a56fb80/ : SUCCESS in 1h 19m 27s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-openshift/eee9f46/ : FAILURE in 1h 40m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized/4bd68a4/ : SUCCESS in 1h 41m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-ovn/83670f8/ : SUCCESS in 1h 38m 05s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-py36/ae7495f/ : SUCCESS in 1h 38m 17s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-py36/e3ee92e/ : SUCCESS in 1h 45m 38s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-multinode-containerized/a32b017/ : FAILURE in 2h 12m 48s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-centos-7/ab03e61/ : SUCCESS in 1h 34m 26s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d8f0b2f1ad10f4bd1f8908b43aa44c9a1f9b155d","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-15 17:27:43.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"13ad3dec8e6b45ea3a07d4e94ecfaa4a77598d0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 18:40:39.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-openshift/c18ca12/ : FAILURE in 1h 00m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-l2/a8fa4d8/ : FAILURE in 1h 06m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/493e82e/ : FAILURE in 54m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-network-policy/d5bb3de/ : FAILURE in 54m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/e06f1ea/ : FAILURE in 1h 04m 28s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-ovn/51289e0/ : FAILURE in 50m 25s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-openshift-ingress/1c471bb/ : FAILURE in 56m 46s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/e6f57e7/ : FAILURE in 50m 08s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-multinode-ha/b53e377/ : FAILURE in 55m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-crio/32ebcea/ : FAILURE in 28m 59s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e7c7ed290e3b261670f965cfb462aa9302ece180","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-17 10:50:47.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"35463047757bd6d43c45ae40d296b839b689ab54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:01:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-openshift/b47e19f/ : FAILURE in 1h 35m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-l2/710855f/ : FAILURE in 1h 54m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/81f09b2/ : TIMED_OUT in 3h 05m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-network-policy/e21f245/ : SUCCESS in 1h 08m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/1cb8461/ : TIMED_OUT in 3h 05m 24s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-ovn/1474042/ : SUCCESS in 1h 32m 47s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-openshift-ingress/a595432/ : FAILURE in 1h 04m 12s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/23944cd/ : FAILURE in 1h 54m 20s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-multinode-ha/b62e00d/ : TIMED_OUT in 2h 15m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-crio/f2217ff/ : FAILURE in 29m 26s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4d11e149869bc695481287c778163160fd6dcdfb","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-17 14:36:47.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"acf3814b9d7df44256d006ffce54bf5b9995b9a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 16:29:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/12/check/openstack-tox-docs/81a3eaa/html/ : SUCCESS in 7m 35s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/12/check/openstack-tox-pep8/5fe463c/ : SUCCESS in 4m 03s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest/13a7fe2/ : SUCCESS in 1h 27m 52s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-openshift/0438758/ : FAILURE in 1h 47m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized/b5bf283/ : SUCCESS in 1h 35m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-ovn/359dc54/ : SUCCESS in 1h 33m 23s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-py36/fa7f119/ : SUCCESS in 1h 06m 30s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-py36/55e302b/ : SUCCESS in 1h 42m 53s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-multinode-containerized/5eb18d1/ : SUCCESS in 1h 36m 57s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-centos-7/855893d/ : FAILURE in 1h 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9c007543ff19e4bd1e80e28002f5bb9e27c2292d","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-18 15:23:38.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"40c495b02976e7e4780c7068bc28ea67347f12c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 17:33:25.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/12/check/openstack-tox-docs/7a008c4/html/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/12/check/openstack-tox-pep8/314fb05/ : SUCCESS in 4m 20s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest/c99756e/ : SUCCESS in 1h 40m 19s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-openshift/b6f32f1/ : FAILURE in 1h 34m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized/5bbea43/ : SUCCESS in 1h 28m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-ovn/d8e0362/ : FAILURE in 1h 53m 23s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-py36/a757fc4/ : FAILURE in 30m 34s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-py36/f23f3c9/ : SUCCESS in 1h 35m 25s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-multinode-containerized/42090c4/ : FAILURE in 1h 43m 19s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-centos-7/a7da033/ : FAILURE in 1h 39m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c78fbda8ee2f74f1f04d9fb1da663df7a8ca2a09","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-19 06:58:47.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"93b91f375ff3de775ea16e052a649afecf4be44b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-19 06:58:53.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"df9fea5bbe3720931026af05cc09d882ade822f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 09:06:32.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/12/check/openstack-tox-docs/b99b7dc/html/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/12/check/openstack-tox-pep8/d020298/ : SUCCESS in 4m 37s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest/f68d71a/ : SUCCESS in 1h 35m 23s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-openshift/0077356/ : FAILURE in 47m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized/ecea83d/ : SUCCESS in 1h 38m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-ovn/b5699ae/ : FAILURE in 1h 44m 44s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-py36/b3bb1b0/ : FAILURE in 1h 03m 27s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-py36/3878d55/ : FAILURE in 1h 09m 16s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-multinode-containerized/8b7d7a6/ : SUCCESS in 1h 47m 52s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-centos-7/0d1055e/ : SUCCESS in 1h 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a4287c40617085b4b508caa2a85aba9c394a4b02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 09:38:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-openshift/243d326/ : FAILURE in 1h 37m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-l2/12b5a69/ : FAILURE in 1h 40m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/48dbe6f/ : FAILURE in 1h 08m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-network-policy/fd73867/ : SUCCESS in 1h 03m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/210c4e9/ : FAILURE in 37m 49s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-ovn/8743253/ : FAILURE in 1h 18m 47s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-openshift-ingress/69bd309/ : FAILURE in 1h 02m 13s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/0d2bfce/ : FAILURE in 1h 33m 53s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-multinode-ha/519916d/ : TIMED_OUT in 2h 15m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-crio/db7d059/ : FAILURE in 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0642bb601dc08186f7cd4f8290b412fd54007189","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-21 08:08:25.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"84da70f12fdc6d977f08b59f05eaae31bd318ab7","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-21 08:08:31.000000000","message":"Patch Set 12:\n\ncheck experimental","accounts_in_message":[],"_revision_number":12},{"id":"a7298a4ce106bc277c8136cd1c4adfb3973dc109","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 11:18:10.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/12/check/openstack-tox-docs/f2df979/html/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/12/check/openstack-tox-pep8/c5ed689/ : SUCCESS in 4m 13s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest/cecb2d9/ : SUCCESS in 2h 22m 02s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-openshift/d067378/ : FAILURE in 2h 31m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized/4d2b4a3/ : SUCCESS in 1h 51m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-lower-constraints/b8d9e16/ : SUCCESS in 1h 53m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-ovn/83626c9/ : SUCCESS in 1h 25m 13s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-py36/17a2af5/ : SUCCESS in 1h 18m 01s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-containerized-py36/5df9c44/ : TIMED_OUT in 3h 04m 20s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-multinode-containerized/bf18cc8/ : SUCCESS in 2h 01m 07s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-centos-7/be598dc/ : SUCCESS in 1h 43m 15s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e023a63ea966bde1aaa8d52ad9d4944fb145333a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 11:20:02.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-openshift/44fc71b/ : FAILURE in 2h 57m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-l2/48621b1/ : FAILURE in 2h 45m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/f400dce/ : TIMED_OUT in 3h 04m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-network-policy/f42db21/ : SUCCESS in 1h 03m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/b840467/ : TIMED_OUT in 3h 04m 15s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-ovn/00302af/ : SUCCESS in 1h 21m 02s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-openshift-ingress/e22186a/ : FAILURE in 59m 54s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/c01d0b8/ : FAILURE in 2h 18m 28s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-multinode-ha/276e341/ : TIMED_OUT in 2h 15m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/12/experimental/kuryr-kubernetes-tempest-containerized-crio/d9b5e15/ : FAILURE in 26m 57s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b12cf12c130296b368a517181dde9f6229a2bf2b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-25 07:38:03.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"36111f345a88d5ab9d122cf67aec6274bfc3ddad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 11:11:58.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/12/check/openstack-tox-docs/d7ec9ce/html/ : SUCCESS in 2m 33s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/12/check/openstack-tox-pep8/eb63a9d/ : SUCCESS in 2m 36s\n- kuryr-kubernetes-tempest finger://ze04.openstack.org/e1b2e5083fb64119a06711e78f16ab20 : RETRY_LIMIT in 50m 19s\n- kuryr-kubernetes-tempest-openshift finger://ze11.openstack.org/b586aa88c09c4ec3b64da0bbda0149b6 : RETRY_LIMIT in 51m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized finger://ze06.openstack.org/a6529d7ca5434b50bb1ecabbd1e79468 : RETRY_LIMIT in 49m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints finger://ze11.openstack.org/7e18a6d31a714687b2b82dafb4bc4451 : RETRY_LIMIT in 51m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn finger://ze09.openstack.org/d4b8a4372a9849fa8ecf44abcb4bd002 : RETRY_LIMIT in 42m 24s (non-voting)\n- kuryr-kubernetes-tempest-py36 finger://ze06.openstack.org/89ae4b27ef07445c8a0a9693a12bdece : RETRY_LIMIT in 42m 38s\n- kuryr-kubernetes-tempest-containerized-py36 finger://ze11.openstack.org/1767d921dc724833b61154593a63ef70 : RETRY_LIMIT in 46m 46s\n- kuryr-kubernetes-tempest-multinode-containerized finger://ze05.openstack.org/fc90145df4ee457a8ed708d5becaa307 : RETRY_LIMIT in 45m 44s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/12/check/kuryr-kubernetes-tempest-centos-7/46028a0/ : FAILURE in 1h 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"03ddd3f96f657d6f4337339ff92683260f8fffaa","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-25 12:22:17.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"d2969b24c69ca39b8b468f41e30c012e78c323f1","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-25 12:23:32.000000000","message":"Patch Set 13:\n\ncheck experimental","accounts_in_message":[],"_revision_number":13},{"id":"c4048beb8b45bf305e42238d5765491c0d31b9ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 19:54:54.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/13/check/openstack-tox-docs/80d8c1e/html/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/13/check/openstack-tox-pep8/21c2b1e/ : SUCCESS in 4m 36s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest/6f1b089/ : SUCCESS in 1h 43m 01s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-openshift/ee8d8b2/ : POST_FAILURE in 59m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized/7c7f42c/ : FAILURE in 1h 36m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-lower-constraints/c78171f/ : SUCCESS in 1h 26m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-ovn/4632198/ : SUCCESS in 1h 36m 40s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-py36/f069756/ : SUCCESS in 1h 31m 59s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-py36/6918558/ : SUCCESS in 1h 39m 27s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-multinode-containerized/6576a8f/ : SUCCESS in 2h 06m 18s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-centos-7/74b80e2/ : SUCCESS in 1h 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0cbd5693c9988a39f7577d5846a528c4f97de641","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-25 20:35:24.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"a5632e06164c4a0d36d1b438ee7d019948dbbbcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 22:40:07.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/13/check/openstack-tox-docs/5b37910/html/ : SUCCESS in 2m 50s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/13/check/openstack-tox-pep8/5d9fea4/ : SUCCESS in 3m 37s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest/8d52858/ : FAILURE in 1h 39m 46s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-openshift/f0fb7a0/ : FAILURE in 1h 40m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized/5712fac/ : SUCCESS in 1h 44m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-lower-constraints/33816fb/ : FAILURE in 1h 16m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-ovn/6a9a6f2/ : SUCCESS in 1h 53m 15s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-py36/2c35471/ : SUCCESS in 1h 23m 08s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-py36/d5da08e/ : FAILURE in 1h 42m 20s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-multinode-containerized/e25f811/ : SUCCESS in 1h 57m 58s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-centos-7/868185c/ : FAILURE in 1h 03m 39s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"d3b19f044962a167e79d7b15c05b5dcf1c122d2b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-26 05:57:27.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"bf9293e2d4a70992b23650de0e50910f80aba435","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-26 05:57:37.000000000","message":"Patch Set 13:\n\ncheck experimental","accounts_in_message":[],"_revision_number":13},{"id":"2c6326175ad169172ec77e5dbed4804c2caf6caf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 08:18:48.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/13/check/openstack-tox-docs/364e81d/html/ : SUCCESS in 3m 02s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/13/check/openstack-tox-pep8/19a9b6b/ : SUCCESS in 3m 01s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest/0786132/ : SUCCESS in 1h 36m 49s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-openshift/3d6daeb/ : SUCCESS in 1h 48m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized/b72110e/ : SUCCESS in 1h 36m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-lower-constraints/718a814/ : SUCCESS in 1h 47m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-ovn/afa3f61/ : SUCCESS in 1h 33m 41s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-py36/c1f698a/ : SUCCESS in 1h 25m 50s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-py36/7f1074f/ : SUCCESS in 1h 29m 42s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-multinode-containerized/9219649/ : SUCCESS in 1h 52m 51s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-centos-7/b4287cf/ : SUCCESS in 1h 52m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c7bd409e931511d15f1afb152f4c4dcbc491b86b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 09:10:41.000000000","message":"Patch Set 13:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-containerized-openshift/f4567d0/ : FAILURE in 2h 01m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-containerized-l2/1ad171a/ : FAILURE in 1h 56m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/00e7eeb/ : TIMED_OUT in 3h 07m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-containerized-network-policy/64e253b/ : SUCCESS in 1h 07m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/753f2e4/ : SUCCESS in 2h 44m 24s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-ovn/207cf64/ : SUCCESS in 1h 25m 03s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-openshift-ingress/a0e3c34/ : FAILURE in 1h 02m 41s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/4d34ff5/ : FAILURE in 1h 33m 15s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-multinode-ha/a2fb380/ : TIMED_OUT in 2h 15m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-containerized-crio/9f37233/ : FAILURE in 37m 05s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"1f48fe99522f12809dec0eb1cd0083094910b616","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-06-26 09:59:13.000000000","message":"Patch Set 13: Code-Review+1\n\nNice! LGTM.","accounts_in_message":[],"_revision_number":13},{"id":"19abe70f6a49b2a4106a473b6b0d1eae8bb9b334","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-06-26 15:31:20.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"23693592d04042661d9e0dddadc31bdaea0afe55","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-26 16:25:42.000000000","message":"Patch Set 13:\n\ncheck experimental","accounts_in_message":[],"_revision_number":13},{"id":"b6aea7ddbe73f6a48700cae77e740cd3b66d1ae1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 19:36:29.000000000","message":"Patch Set 13:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-containerized-openshift/f40d1df/ : FAILURE in 1h 32m 25s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-containerized-l2/415f08b/ : FAILURE in 1h 31m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/b31b296/ : TIMED_OUT in 3h 04m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-containerized-network-policy/55003bb/ : SUCCESS in 1h 09m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/7fafb37/ : SUCCESS in 2h 43m 29s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-ovn/1e50ab1/ : SUCCESS in 1h 53m 35s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-openshift-ingress/06294ae/ : FAILURE in 1h 02m 34s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/5541e09/ : FAILURE in 2h 01m 48s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-multinode-ha/a226364/ : TIMED_OUT in 2h 16m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/13/experimental/kuryr-kubernetes-tempest-containerized-crio/0a97cd4/ : FAILURE in 26m 30s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f291cc34819eb1f8f7a3c5c8f1b8c9b4861c006f","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-27 05:55:25.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"e98d9cc16fc5f2e768b269bb9f4d680379319a7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 09:06:22.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/13/check/openstack-tox-docs/53c58b6/html/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/13/check/openstack-tox-pep8/614fbf9/ : SUCCESS in 3m 01s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest/820a998/ : SUCCESS in 1h 32m 21s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-openshift/49a4d64/ : FAILURE in 1h 40m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized/ab806cd/ : SUCCESS in 1h 38m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-lower-constraints/3868343/ : FAILURE in 1h 27m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-ovn/7dfd61b/ : FAILURE in 1h 29m 27s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-py36/58ce032/ : FAILURE in 1h 22m 10s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-py36/96ad6c2/ : TIMED_OUT in 3h 03m 57s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-multinode-containerized/21c1aef/ : SUCCESS in 1h 57m 49s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-centos-7/7c98d98/ : SUCCESS in 1h 39m 52s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"608fd6ab38dfddb487067b603e023d94c1ae10c1","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-06-27 10:39:20.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"622f28796f6539c1a968ed195ccb54ce3c4088b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:00:56.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/13/check/openstack-tox-docs/b72690c/html/ : SUCCESS in 2m 52s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/13/check/openstack-tox-pep8/7c788f6/ : SUCCESS in 3m 11s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest/4ac9fcf/ : TIMED_OUT in 3h 07m 42s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-openshift/7abcbc5/ : FAILURE in 1h 58m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized/f347043/ : SUCCESS in 1h 41m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-lower-constraints/fad6562/ : SUCCESS in 1h 23m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-ovn/4296628/ : SUCCESS in 1h 35m 45s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-py36/946ae6b/ : FAILURE in 1h 59m 59s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-py36/3b04277/ : SUCCESS in 2h 13m 09s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-multinode-containerized/512c05d/ : SUCCESS in 1h 46m 05s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-centos-7/ec3c427/ : SUCCESS in 1h 31m 51s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"db19abda918a531ae3d03a312f781fd0297e7a05","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-06-27 14:34:23.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"5f4fe3711bccc836fe3dcd76c7947831bffc5dc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 16:55:45.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/13/check/openstack-tox-docs/082d0d2/html/ : SUCCESS in 2m 47s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/13/check/openstack-tox-pep8/9c78a83/ : SUCCESS in 3m 27s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest/d4c0bb3/ : SUCCESS in 1h 21m 16s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-openshift/23734ca/ : FAILURE in 1h 49m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized/c210e1f/ : SUCCESS in 1h 31m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-lower-constraints/76993ed/ : SUCCESS in 1h 40m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-ovn/be31df4/ : FAILURE in 14m 23s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-py36/5d4e8c7/ : SUCCESS in 1h 14m 56s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-containerized-py36/3d127aa/ : SUCCESS in 1h 48m 52s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-multinode-containerized/13760df/ : SUCCESS in 2h 11m 50s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/13/check/kuryr-kubernetes-tempest-centos-7/6766320/ : SUCCESS in 1h 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c37aab48cf4e13f414a68b7ce36530c77203168b","author":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"date":"2019-07-01 10:39:01.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"9dd2f164a58b78d091fd1fbcee6b732df1ff835e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-07-01 10:54:36.000000000","message":"Patch Set 13: Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"2393051430f57b5d636c10bcf3512dcad7a96fbb","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-04 08:03:34.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"c3c313392fc78f3ff085acca51db9f6cbd704dd8","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-07-04 08:04:46.000000000","message":"Patch Set 14: Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"70e419bad2b36371c222a4e3a8171c1ca66e6b09","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-04 08:05:23.000000000","message":"Patch Set 14:\n\ncheck experimental","accounts_in_message":[],"_revision_number":14},{"id":"53047411294781a112010e776d0a50fb8cadf7ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:22:05.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/14/check/openstack-tox-docs/801a76d/html/ : SUCCESS in 2m 57s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/14/check/openstack-tox-pep8/1e21a18/ : SUCCESS in 2m 54s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest/3468e26/ : FAILURE in 37m 40s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-openshift/c76018b/ : FAILURE in 1h 37m 47s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-containerized/df147cc/ : SUCCESS in 1h 38m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-containerized-lower-constraints/53b33b9/ : SUCCESS in 1h 22m 20s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-containerized-ovn/98e5aa5/ : SUCCESS in 1h 38m 09s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-py36/1ea9811/ : SUCCESS in 1h 25m 29s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-containerized-py36/433b4df/ : SUCCESS in 1h 38m 16s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-multinode-containerized/dd65d3c/ : SUCCESS in 1h 56m 55s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-centos-7/8f84c5d/ : SUCCESS in 1h 53m 57s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6e5eab2a8bbf3788f5e7edee04b565550fd92326","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-04 10:36:00.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"981402c9ddba191a1f38a007ddcc13ad0e07d50a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 11:46:37.000000000","message":"Patch Set 14:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/14/experimental/kuryr-kubernetes-tempest-containerized-openshift/5aa64f2/ : FAILURE in 1h 15m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/14/experimental/kuryr-kubernetes-tempest-containerized-l2/b6c8022/ : FAILURE in 1h 48m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/14/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/1deddec/ : TIMED_OUT in 3h 05m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/14/experimental/kuryr-kubernetes-tempest-containerized-network-policy/1244242/ : SUCCESS in 56m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/14/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/acd1137/ : FAILURE in 1h 55m 24s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/14/experimental/kuryr-kubernetes-tempest-ovn/aa74521/ : SUCCESS in 1h 41m 47s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/14/experimental/kuryr-kubernetes-tempest-openshift-ingress/7c9a6ad/ : FAILURE in 1h 01m 42s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/14/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/d0ff126/ : FAILURE in 1h 52m 15s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/14/experimental/kuryr-kubernetes-tempest-multinode-ha/694c6e1/ : TIMED_OUT in 2h 15m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/14/experimental/kuryr-kubernetes-tempest-containerized-crio/009091a/ : FAILURE in 31m 37s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"14f9d25e20a34de68fce8a929265f1bc70a307c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 12:50:45.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/14/check/openstack-tox-docs/bac5353/html/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/14/check/openstack-tox-pep8/f7895c4/ : SUCCESS in 3m 23s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest/123a8e6/ : SUCCESS in 1h 51m 16s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-openshift/506b53f/ : FAILURE in 1h 34m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-containerized/a391929/ : SUCCESS in 1h 55m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-containerized-lower-constraints/4df4cf3/ : SUCCESS in 1h 53m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-containerized-ovn/ff93d87/ : SUCCESS in 1h 52m 26s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-py36/a70a95a/ : SUCCESS in 1h 36m 59s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-containerized-py36/d1b18a1/ : SUCCESS in 1h 42m 13s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-multinode-containerized/cdf71b3/ : SUCCESS in 2h 02m 44s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/14/check/kuryr-kubernetes-tempest-centos-7/594cf52/ : SUCCESS in 2h 08m 28s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e56ce8243c670dc8c8da83310e87b0d5a4f57a11","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-09 12:03:40.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"96c04ad2c1f6bc160c234d71ae4ae1d286491d04","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-09 12:04:44.000000000","message":"Patch Set 15: Workflow+1\n\njust triggered the rebase. Readding the +W","accounts_in_message":[],"_revision_number":15},{"id":"7f7bc41852f98688b2ff79c113efe7cfbe3ecc81","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-07-09 12:05:20.000000000","message":"Patch Set 15:\n\ncheck experimental","accounts_in_message":[],"_revision_number":15},{"id":"7bf80016998ef8cee69664b6669d12914e5aa01b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 14:05:51.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/15/check/openstack-tox-docs/5a1c16b/html/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/15/check/openstack-tox-pep8/fe11166/ : SUCCESS in 2m 58s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/15/check/kuryr-kubernetes-tempest/b39b0f4/ : SUCCESS in 1h 12m 21s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/51/618751/15/check/kuryr-kubernetes-tempest-openshift/3cf5dd3/ : FAILURE in 1h 38m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/51/618751/15/check/kuryr-kubernetes-tempest-containerized/5ece918/ : SUCCESS in 1h 40m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-lower-constraints http://logs.openstack.org/51/618751/15/check/kuryr-kubernetes-tempest-containerized-lower-constraints/5be7882/ : SUCCESS in 1h 33m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/51/618751/15/check/kuryr-kubernetes-tempest-containerized-ovn/796b669/ : SUCCESS in 1h 46m 13s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/15/check/kuryr-kubernetes-tempest-py36/0a377bd/ : SUCCESS in 1h 18m 15s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/15/check/kuryr-kubernetes-tempest-containerized-py36/3a72730/ : SUCCESS in 1h 15m 04s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/51/618751/15/check/kuryr-kubernetes-tempest-multinode-containerized/fd81a7d/ : SUCCESS in 1h 39m 19s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/51/618751/15/check/kuryr-kubernetes-tempest-centos-7/f551288/ : SUCCESS in 1h 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0448f9b42a976d57fc584f54f6c3c72f2bfea7d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 14:06:04.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"a5c80d58a09129381b627bbfb00617a34ea7fa51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 14:46:50.000000000","message":"Patch Set 15:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/51/618751/15/experimental/kuryr-kubernetes-tempest-containerized-openshift/8ce3e35/ : FAILURE in 1h 44m 57s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/51/618751/15/experimental/kuryr-kubernetes-tempest-containerized-l2/aea19aa/ : FAILURE in 1h 51m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/51/618751/15/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/09903e9/ : FAILURE in 1h 56m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/51/618751/15/experimental/kuryr-kubernetes-tempest-containerized-network-policy/601d554/ : SUCCESS in 1h 05m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/51/618751/15/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/1a07774/ : FAILURE in 2h 35m 07s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/51/618751/15/experimental/kuryr-kubernetes-tempest-ovn/6671992/ : SUCCESS in 1h 25m 48s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/51/618751/15/experimental/kuryr-kubernetes-tempest-openshift-ingress/a343d1a/ : FAILURE in 2h 18m 38s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/51/618751/15/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/05ea6b7/ : FAILURE in 1h 52m 09s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/51/618751/15/experimental/kuryr-kubernetes-tempest-multinode-ha/c8cca17/ : TIMED_OUT in 2h 19m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/51/618751/15/experimental/kuryr-kubernetes-tempest-containerized-crio/6e5730b/ : FAILURE in 30m 53s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"255961d07cc41c19b679ff538f6711cde1d32467","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 15:50:24.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/51/618751/15/gate/openstack-tox-docs/93a07ac/html/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/51/618751/15/gate/openstack-tox-pep8/3d94e4e/ : SUCCESS in 3m 28s\n- kuryr-kubernetes-tempest http://logs.openstack.org/51/618751/15/gate/kuryr-kubernetes-tempest/14111e9/ : SUCCESS in 1h 21m 00s\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/51/618751/15/gate/kuryr-kubernetes-tempest-py36/c245e73/ : SUCCESS in 1h 27m 49s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/51/618751/15/gate/kuryr-kubernetes-tempest-containerized-py36/b0574f5/ : SUCCESS in 1h 36m 14s","accounts_in_message":[],"_revision_number":15},{"id":"8981a13278554ec2829e2f21c97f026e443a91d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 15:50:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"ad648cd40614cf1ac9836f39ef6e4a7a0edbdaa6","revisions":{"75c022c2ecf0f26e4b087736ce40861108e45e84":{"kind":"REWORK","_number":1,"created":"2018-11-19 13:42:36.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/1"}}},"commit":{"parents":[{"commit":"a27fe3d35f1fd1ac2b5d18a21575a67487a5c53c","subject":"Add timeout parameter to \u0027connect_get_namespaced_pod_exec\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a27fe3d35f1fd1ac2b5d18a21575a67487a5c53c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:42:14.000000000","tz":60},"subject":"Enhance svc namespace isolation with LoadBalancer type svc","message":"Enhance svc namespace isolation with LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/75c022c2ecf0f26e4b087736ce40861108e45e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/75c022c2ecf0f26e4b087736ce40861108e45e84"}]},"branch":"refs/heads/master"},"7996d4177f4165f104731b69c7387c2f2623d921":{"kind":"REWORK","_number":2,"created":"2018-11-20 09:58:34.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/2"}}},"commit":{"parents":[{"commit":"a27fe3d35f1fd1ac2b5d18a21575a67487a5c53c","subject":"Add timeout parameter to \u0027connect_get_namespaced_pod_exec\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a27fe3d35f1fd1ac2b5d18a21575a67487a5c53c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-20 09:57:51.000000000","tz":60},"subject":"Enhance svc namespace isolation with LoadBalancer type svc","message":"Enhance svc namespace isolation with LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7996d4177f4165f104731b69c7387c2f2623d921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7996d4177f4165f104731b69c7387c2f2623d921"}]},"branch":"refs/heads/master"},"5c18a3588ff09ac377041005f5e8b0f16e35b3f5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-11-30 11:19:51.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/3"}}},"commit":{"parents":[{"commit":"d8d626738f97dc7a90bfab99d1b5804cb46cc4e8","subject":"Merge \"Check the number of neutron ports after controller restart\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d8d626738f97dc7a90bfab99d1b5804cb46cc4e8"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-30 11:19:51.000000000","tz":0},"subject":"Enhance svc namespace isolation with LoadBalancer type svc","message":"Enhance svc namespace isolation with LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5c18a3588ff09ac377041005f5e8b0f16e35b3f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5c18a3588ff09ac377041005f5e8b0f16e35b3f5"}]},"branch":"refs/heads/master"},"0a2b5c1647451ebfd7965e49ebb5c2ebc1942ad0":{"kind":"REWORK","_number":4,"created":"2018-11-30 15:48:41.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/4"}}},"commit":{"parents":[{"commit":"a27fe3d35f1fd1ac2b5d18a21575a67487a5c53c","subject":"Add timeout parameter to \u0027connect_get_namespaced_pod_exec\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a27fe3d35f1fd1ac2b5d18a21575a67487a5c53c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-30 15:48:29.000000000","tz":60},"subject":"Enhance svc namespace isolation with LoadBalancer type svc","message":"Enhance svc namespace isolation with LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0a2b5c1647451ebfd7965e49ebb5c2ebc1942ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0a2b5c1647451ebfd7965e49ebb5c2ebc1942ad0"}]},"branch":"refs/heads/master"},"ddcc10056cb649649f5fbacea48b9a74767bddb7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2018-12-07 08:56:28.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/5"}}},"commit":{"parents":[{"commit":"aa6c406e399960887f8af265913782153ae85510","subject":"Merge \"Add possibility to build with podman/buildah\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/aa6c406e399960887f8af265913782153ae85510"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-12-07 08:56:28.000000000","tz":0},"subject":"Enhance svc namespace isolation with LoadBalancer type svc","message":"Enhance svc namespace isolation with LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ddcc10056cb649649f5fbacea48b9a74767bddb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ddcc10056cb649649f5fbacea48b9a74767bddb7"}]},"branch":"refs/heads/master"},"468fb9ed0e2c82b129328798434ad9233e4269c7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2018-12-10 08:03:54.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/6"}}},"commit":{"parents":[{"commit":"8ec2c709a2a51a7a6be06c1c9f27a305bf8eba6c","subject":"Synchronize port pool tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8ec2c709a2a51a7a6be06c1c9f27a305bf8eba6c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-12-10 08:03:54.000000000","tz":0},"subject":"Enhance svc namespace isolation with LoadBalancer type svc","message":"Enhance svc namespace isolation with LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/468fb9ed0e2c82b129328798434ad9233e4269c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/468fb9ed0e2c82b129328798434ad9233e4269c7"}]},"branch":"refs/heads/master"},"51bdeba5d2e89d62ec768c13f35b3771dd5eca82":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2018-12-24 13:35:48.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/7"}}},"commit":{"parents":[{"commit":"7f84979019b4cfc44927bd494c4c37985e6aa209","subject":"Merge \"Making sure all the pods reside on the same node for port_pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7f84979019b4cfc44927bd494c4c37985e6aa209"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-12-24 13:35:48.000000000","tz":0},"subject":"Enhance svc namespace isolation with LoadBalancer type svc","message":"Enhance svc namespace isolation with LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/51bdeba5d2e89d62ec768c13f35b3771dd5eca82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/51bdeba5d2e89d62ec768c13f35b3771dd5eca82"}]},"branch":"refs/heads/master"},"df29fce4e7c269d56d412cd1f6808ba2abde00f7":{"kind":"REWORK","_number":8,"created":"2019-05-24 10:13:28.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/8"}}},"commit":{"parents":[{"commit":"c9ef83d0fb6335a8728a8c2fe796d2f30168295b","subject":"Create/delete network policy in non-default namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c9ef83d0fb6335a8728a8c2fe796d2f30168295b"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-05-24 10:12:20.000000000","tz":120},"subject":"Enhance svc namespace isolation with LoadBalancer type svc","message":"Enhance svc namespace isolation with LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/df29fce4e7c269d56d412cd1f6808ba2abde00f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/df29fce4e7c269d56d412cd1f6808ba2abde00f7"}]},"branch":"refs/heads/master"},"14314be07427081b668b4bff3fe0bb73212123ee":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-06-04 06:34:47.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/9"}}},"commit":{"parents":[{"commit":"6c4ac2bebd235264f0787006d3478032dee450ea","subject":"detect failed curl when streamed from pod","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6c4ac2bebd235264f0787006d3478032dee450ea"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-04 06:34:47.000000000","tz":0},"subject":"Enhance svc namespace isolation with LoadBalancer type svc","message":"Enhance svc namespace isolation with LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/14314be07427081b668b4bff3fe0bb73212123ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/14314be07427081b668b4bff3fe0bb73212123ee"}]},"branch":"refs/heads/master"},"2f3e9cd4d1642ac75738a7f89f9ed30e286ecb5b":{"kind":"REWORK","_number":10,"created":"2019-06-11 07:07:48.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/10"}}},"commit":{"parents":[{"commit":"d92ef7013ede60ada070cd7ea200d872460fb85b","subject":"Remove sleep time in favor of checking members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d92ef7013ede60ada070cd7ea200d872460fb85b"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-11 07:07:03.000000000","tz":120},"subject":"Enhance svc namespace isolation with LoadBalancer type svc","message":"Enhance svc namespace isolation with LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2f3e9cd4d1642ac75738a7f89f9ed30e286ecb5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2f3e9cd4d1642ac75738a7f89f9ed30e286ecb5b"}]},"branch":"refs/heads/master"},"1e699428df3912d566034756ba5991711221412a":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2019-06-11 10:25:56.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/11"}}},"commit":{"parents":[{"commit":"d92ef7013ede60ada070cd7ea200d872460fb85b","subject":"Remove sleep time in favor of checking members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d92ef7013ede60ada070cd7ea200d872460fb85b"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-11 10:25:45.000000000","tz":0},"subject":"Enhance svc namespace isolation for LoadBalancer type svc","message":"Enhance svc namespace isolation for LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1e699428df3912d566034756ba5991711221412a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/1e699428df3912d566034756ba5991711221412a"}]},"branch":"refs/heads/master"},"b59bf4f6fb00ab09e03952f81ca3144ed2676c9b":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2019-06-13 07:09:40.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/12"}}},"commit":{"parents":[{"commit":"6fe8fc5b9fe86a366fca67fcb2210e2df486e22f","subject":"Remove sleep time in favor of checking members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6fe8fc5b9fe86a366fca67fcb2210e2df486e22f"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-13 07:09:13.000000000","tz":120},"subject":"Enhance svc namespace isolation for LoadBalancer type svc","message":"Enhance svc namespace isolation for LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nDepends-On: I35f3be8fc16d71d0d2ac01c5451d398dd631f118\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b59bf4f6fb00ab09e03952f81ca3144ed2676c9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b59bf4f6fb00ab09e03952f81ca3144ed2676c9b"}]},"branch":"refs/heads/master"},"c9336dc686ced587a4d7a0be1c842d474938bff8":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-06-25 12:22:17.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/13"}}},"commit":{"parents":[{"commit":"ca12210bb3c37324389e9222b16729ffe5128cb3","subject":"Fixing service connectivity testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ca12210bb3c37324389e9222b16729ffe5128cb3"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-25 12:21:49.000000000","tz":120},"subject":"Enhance svc namespace isolation for LoadBalancer type svc","message":"Enhance svc namespace isolation for LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nDepends-On: I35f3be8fc16d71d0d2ac01c5451d398dd631f118\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c9336dc686ced587a4d7a0be1c842d474938bff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c9336dc686ced587a4d7a0be1c842d474938bff8"}]},"branch":"refs/heads/master"},"f2619d13eafdb827d8b5350f389aac3afe095eaf":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-07-04 08:03:34.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/14"}}},"commit":{"parents":[{"commit":"0e6a970b607a61a140a8d12b014e65874cdd696b","subject":"Fixing service connectivity testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0e6a970b607a61a140a8d12b014e65874cdd696b"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-07-04 08:02:42.000000000","tz":120},"subject":"Enhance svc namespace isolation for LoadBalancer type svc","message":"Enhance svc namespace isolation for LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nDepends-On: I35f3be8fc16d71d0d2ac01c5451d398dd631f118\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f2619d13eafdb827d8b5350f389aac3afe095eaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f2619d13eafdb827d8b5350f389aac3afe095eaf"}]},"branch":"refs/heads/master"},"ad648cd40614cf1ac9836f39ef6e4a7a0edbdaa6":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-07-09 12:03:40.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/51/618751/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/51/618751/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/51/618751/15"}}},"commit":{"parents":[{"commit":"54220e17fa0a32036fd35b981a9dd1c06997823c","subject":"Fixing service connectivity testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/54220e17fa0a32036fd35b981a9dd1c06997823c"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2018-11-19 13:40:32.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-07-09 12:03:40.000000000","tz":0},"subject":"Enhance svc namespace isolation for LoadBalancer type svc","message":"Enhance svc namespace isolation for LoadBalancer type svc\n\nThis patch extends the current service isolation test with a service\nof LoadBalancer type, checking that it has connectivity across\nnamespaces depite of the namespace isolation.\n\nDepends-On: I35f3be8fc16d71d0d2ac01c5451d398dd631f118\n\nChange-Id: I40bf39c4c1d0b43f5d83e6aed81786390753ce99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ad648cd40614cf1ac9836f39ef6e4a7a0edbdaa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ad648cd40614cf1ac9836f39ef6e4a7a0edbdaa6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
