)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~626892","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ia2be9260c31fcde0da508eb3d7021d69b99f27d2","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"proper_error_stream_pod","hashtags":[],"change_id":"Ia2be9260c31fcde0da508eb3d7021d69b99f27d2","subject":"detect failed curl when streamed from pod","status":"MERGED","created":"2018-12-21 13:21:41.000000000","updated":"2019-05-28 13:48:57.000000000","submitted":"2019-05-28 13:48:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"626892-1559051337711-f0e816bf","meta_rev_id":"c4fc07be87db7057d40bb6f321766db645be4708","_number":626892,"virtual_id_number":626892,"owner":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":2,"date":"2019-05-28 13:48:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2019-05-28 09:56:08.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2019-05-28 12:24:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-28 10:02:01.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-28 10:02:01.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-28 09:56:08.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-05-28 10:02:01.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-05-28 12:24:54.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2019-05-28 13:48:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97c61aa151e425517b68f83d367a2da2f69a7c3f","author":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"date":"2018-12-21 13:21:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e225693135006c28f107e9f0f8a5bc188788429b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-21 14:51:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/1/check/openstack-tox-docs/b2810b6/html/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/1/check/openstack-tox-pep8/0a6f986/ : FAILURE in 4m 00s\n- kuryr-kubernetes-tempest-octavia http://logs.openstack.org/92/626892/1/check/kuryr-kubernetes-tempest-octavia/fa88893/ : FAILURE in 1h 03m 37s\n- kuryr-kubernetes-tempest-daemon-octavia http://logs.openstack.org/92/626892/1/check/kuryr-kubernetes-tempest-daemon-octavia/cb8f348/ : FAILURE in 50m 47s\n- kuryr-kubernetes-tempest-daemon-openshift-octavia http://logs.openstack.org/92/626892/1/check/kuryr-kubernetes-tempest-daemon-openshift-octavia/2e6a416/ : FAILURE in 25m 50s (non-voting)\n- kuryr-kubernetes-tempest-daemon-lbaasv2 http://logs.openstack.org/92/626892/1/check/kuryr-kubernetes-tempest-daemon-lbaasv2/ab6ee3a/ : FAILURE in 41m 55s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia http://logs.openstack.org/92/626892/1/check/kuryr-kubernetes-tempest-daemon-containerized-octavia/93a8e58/ : FAILURE in 1h 16m 03s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-ovn http://logs.openstack.org/92/626892/1/check/kuryr-kubernetes-tempest-daemon-containerized-ovn/d24f8f2/ : FAILURE in 1h 08m 56s (non-voting)\n- kuryr-kubernetes-tempest-daemon-octavia-py36 http://logs.openstack.org/92/626892/1/check/kuryr-kubernetes-tempest-daemon-octavia-py36/e850273/ : FAILURE in 1h 06m 34s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-py36 http://logs.openstack.org/92/626892/1/check/kuryr-kubernetes-tempest-daemon-containerized-octavia-py36/dd7b074/ : FAILURE in 1h 22m 09s\n- kuryr-kubernetes-tempest-multinode-daemon-octavia-containerized http://logs.openstack.org/92/626892/1/check/kuryr-kubernetes-tempest-multinode-daemon-octavia-containerized/5290195/ : FAILURE in 43m 30s (non-voting)\n- kuryr-kubernetes-tempest-octavia-centos-7 http://logs.openstack.org/92/626892/1/check/kuryr-kubernetes-tempest-octavia-centos-7/cb423af/ : FAILURE in 1h 17m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8e90ce5dfd9b4592fa3b72cdff2b02601044ffd2","author":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"date":"2018-12-24 08:42:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4b635d8e194c10de187116f8382056e208bffe39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-24 10:42:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/2/check/openstack-tox-docs/d073425/html/ : SUCCESS in 4m 14s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/2/check/openstack-tox-pep8/f33f9ef/ : SUCCESS in 4m 05s\n- kuryr-kubernetes-tempest-octavia http://logs.openstack.org/92/626892/2/check/kuryr-kubernetes-tempest-octavia/1de82c6/ : FAILURE in 1h 02m 17s\n- kuryr-kubernetes-tempest-daemon-octavia http://logs.openstack.org/92/626892/2/check/kuryr-kubernetes-tempest-daemon-octavia/2307b37/ : FAILURE in 1h 00m 49s\n- kuryr-kubernetes-tempest-daemon-openshift-octavia http://logs.openstack.org/92/626892/2/check/kuryr-kubernetes-tempest-daemon-openshift-octavia/f115281/ : FAILURE in 1h 02m 18s (non-voting)\n- kuryr-kubernetes-tempest-daemon-lbaasv2 http://logs.openstack.org/92/626892/2/check/kuryr-kubernetes-tempest-daemon-lbaasv2/7a94dfa/ : FAILURE in 1h 55m 47s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia http://logs.openstack.org/92/626892/2/check/kuryr-kubernetes-tempest-daemon-containerized-octavia/3963b56/ : FAILURE in 39m 22s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-ovn http://logs.openstack.org/92/626892/2/check/kuryr-kubernetes-tempest-daemon-containerized-ovn/5bdc470/ : FAILURE in 1h 22m 02s (non-voting)\n- kuryr-kubernetes-tempest-daemon-octavia-py36 http://logs.openstack.org/92/626892/2/check/kuryr-kubernetes-tempest-daemon-octavia-py36/fadee90/ : FAILURE in 56m 16s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-py36 http://logs.openstack.org/92/626892/2/check/kuryr-kubernetes-tempest-daemon-containerized-octavia-py36/cd1e3a8/ : FAILURE in 1h 01m 19s\n- kuryr-kubernetes-tempest-multinode-daemon-octavia-containerized http://logs.openstack.org/92/626892/2/check/kuryr-kubernetes-tempest-multinode-daemon-octavia-containerized/0a604c9/ : FAILURE in 1h 07m 19s (non-voting)\n- kuryr-kubernetes-tempest-octavia-centos-7 http://logs.openstack.org/92/626892/2/check/kuryr-kubernetes-tempest-octavia-centos-7/d447815/ : FAILURE in 1h 24m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d83c5aa9c0d53076d23a1c7fd467b6e504b7ff9d","author":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"date":"2019-01-07 10:16:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f26679418aeb9114773bb750736e2728a7ab5bc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-07 11:50:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/3/check/openstack-tox-docs/925c8ba/html/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/3/check/openstack-tox-pep8/71bdc97/ : SUCCESS in 4m 27s\n- kuryr-kubernetes-tempest-octavia http://logs.openstack.org/92/626892/3/check/kuryr-kubernetes-tempest-octavia/167af89/ : FAILURE in 1h 11m 21s\n- kuryr-kubernetes-tempest-daemon-octavia http://logs.openstack.org/92/626892/3/check/kuryr-kubernetes-tempest-daemon-octavia/9d36637/ : FAILURE in 57m 26s\n- kuryr-kubernetes-tempest-daemon-openshift-octavia http://logs.openstack.org/92/626892/3/check/kuryr-kubernetes-tempest-daemon-openshift-octavia/3d0437f/ : FAILURE in 1h 04m 43s (non-voting)\n- kuryr-kubernetes-tempest-daemon-lbaasv2 http://logs.openstack.org/92/626892/3/check/kuryr-kubernetes-tempest-daemon-lbaasv2/6419b12/ : FAILURE in 37m 18s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia http://logs.openstack.org/92/626892/3/check/kuryr-kubernetes-tempest-daemon-containerized-octavia/7d10ca9/ : FAILURE in 1h 05m 47s (non-voting)\n- kuryr-kubernetes-tempest-daemon-containerized-ovn http://logs.openstack.org/92/626892/3/check/kuryr-kubernetes-tempest-daemon-containerized-ovn/61b3ff0/ : FAILURE in 1h 25m 37s (non-voting)\n- kuryr-kubernetes-tempest-daemon-octavia-py36 http://logs.openstack.org/92/626892/3/check/kuryr-kubernetes-tempest-daemon-octavia-py36/7d5b240/ : FAILURE in 1h 00m 24s\n- kuryr-kubernetes-tempest-daemon-containerized-octavia-py36 http://logs.openstack.org/92/626892/3/check/kuryr-kubernetes-tempest-daemon-containerized-octavia-py36/f399cd2/ : FAILURE in 1h 22m 53s\n- kuryr-kubernetes-tempest-multinode-daemon-octavia-containerized http://logs.openstack.org/92/626892/3/check/kuryr-kubernetes-tempest-multinode-daemon-octavia-containerized/accbe82/ : FAILURE in 1h 26m 33s (non-voting)\n- kuryr-kubernetes-tempest-octavia-centos-7 http://logs.openstack.org/92/626892/3/check/kuryr-kubernetes-tempest-octavia-centos-7/776090e/ : FAILURE in 1h 02m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"49f3ba967f0ca86c38444e6644a71d124781d66a","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-03-14 08:10:47.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"632b8c78fa9168529122b338cf7dec7d0a84d352","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-14 09:57:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/4/check/openstack-tox-docs/9d369e5/html/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/4/check/openstack-tox-pep8/b9cd4ac/ : SUCCESS in 4m 07s\n- kuryr-kubernetes-tempest http://logs.openstack.org/92/626892/4/check/kuryr-kubernetes-tempest/20b5158/ : FAILURE in 1h 25m 23s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/92/626892/4/check/kuryr-kubernetes-tempest-openshift/a3be7f9/ : FAILURE in 1h 19m 27s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/92/626892/4/check/kuryr-kubernetes-tempest-containerized/2cb8905/ : FAILURE in 1h 16m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/92/626892/4/check/kuryr-kubernetes-tempest-containerized-ovn/1a1541b/ : FAILURE in 1h 21m 21s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/92/626892/4/check/kuryr-kubernetes-tempest-py36/3097f75/ : FAILURE in 1h 05m 29s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/92/626892/4/check/kuryr-kubernetes-tempest-containerized-py36/f89f1c1/ : FAILURE in 1h 13m 13s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/92/626892/4/check/kuryr-kubernetes-tempest-multinode-containerized/97895be/ : FAILURE in 1h 42m 52s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/92/626892/4/check/kuryr-kubernetes-tempest-centos-7/71b9862/ : FAILURE in 1h 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"40ab407bfdb706742e13d468475527f9bc6995c7","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-03-31 13:19:59.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e81eb930795d82e0d5d52b6744b3e68549369f50","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-03-31 13:20:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4d1d66be59618b235c676da35b7618c7474de6f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-31 15:32:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/5/check/openstack-tox-docs/3a33df4/html/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/5/check/openstack-tox-pep8/4115dfb/ : SUCCESS in 4m 33s\n- kuryr-kubernetes-tempest http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest/24ca772/ : FAILURE in 59m 25s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-openshift/7d6a10e/ : FAILURE in 1h 01m 38s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized/97e0cf0/ : TIMED_OUT in 2h 04m 12s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-ovn/3fa5817/ : TIMED_OUT in 2h 04m 05s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-py36/545eb93/ : FAILURE in 58m 26s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-py36/d941458/ : TIMED_OUT in 2h 03m 59s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-multinode-containerized/07542d2/ : TIMED_OUT in 2h 04m 18s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-centos-7/4124e9f/ : FAILURE in 1h 07m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5645228ae79c0a64448b9bd34e1f9e60cf4e3314","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-04-08 07:18:08.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"12fad24c9f522e53567dd58c038efd2714463fc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 10:27:59.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/5/check/openstack-tox-docs/532bd6a/ : FAILURE in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/5/check/openstack-tox-pep8/890e9ac/ : SUCCESS in 4m 21s\n- kuryr-kubernetes-tempest http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest/8a99e0f/ : SUCCESS in 1h 35m 30s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-openshift/bf5884a/ : FAILURE in 2h 16m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized/80ee119/ : SUCCESS in 1h 40m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-ovn/695d607/ : TIMED_OUT in 3h 04m 03s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-py36/8cb7e24/ : SUCCESS in 1h 22m 39s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-py36/80c6c1c/ : SUCCESS in 1h 35m 50s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-multinode-containerized/463ebe5/ : SUCCESS in 1h 57m 39s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-centos-7/0c87e94/ : SUCCESS in 1h 40m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"475e616d690223135b17800c1b661424ad75d165","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-10 14:26:53.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ec06d23407ae273d4146de5fbd88e99bde0dabcf","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-10 14:27:02.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"43f4078d78e3f78b7fd9d5ee713abf4e8fa996ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 17:39:14.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/5/check/openstack-tox-docs/519124c/ : FAILURE in 6m 52s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/5/check/openstack-tox-pep8/2b3efed/ : SUCCESS in 7m 21s\n- kuryr-kubernetes-tempest http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest/af0fade/ : SUCCESS in 1h 19m 54s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-openshift/f836fa3/ : SUCCESS in 2h 04m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized/9c26daf/ : SUCCESS in 1h 33m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-ovn/8bcfc54/ : FAILURE in 1h 45m 41s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-py36/7d878e6/ : SUCCESS in 1h 27m 57s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-py36/56a5371/ : TIMED_OUT in 3h 04m 33s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-multinode-containerized/205e9bd/ : SUCCESS in 1h 54m 57s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-centos-7/e6ae5b7/ : SUCCESS in 1h 43m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0c5d6a3e032ebed90cd86f7f5c6cc5ee877ab5eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 17:39:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-openshift/eb15c23/ : TIMED_OUT in 3h 05m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-l2/d3832ec/ : FAILURE in 1h 39m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/9a4b702/ : TIMED_OUT in 3h 05m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-network-policy/a06ec02/ : FAILURE in 1h 25m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/cc7c915/ : TIMED_OUT in 3h 06m 01s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-ovn/fe8b340/ : FAILURE in 2h 03m 32s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-openshift-ingress/dbc603f/ : FAILURE in 1h 07m 59s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/557f566/ : FAILURE in 2h 13m 51s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-multinode-ha/9fdb248/ : TIMED_OUT in 2h 15m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-crio/eeb4d14/ : FAILURE in 25m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b92537665048ba7e01ab0956b7fd7484d55e4927","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-14 08:54:24.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"32275be07a6da08a4c5eadcf01a1a3668ba0405a","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-14 08:54:33.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"42d3e1c2accb254ec70e0aec517b780d37fb9940","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-14 10:56:49.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/5/check/openstack-tox-docs/f513225/ : FAILURE in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/5/check/openstack-tox-pep8/01671d6/ : SUCCESS in 4m 13s\n- kuryr-kubernetes-tempest http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest/0acf7d5/ : SUCCESS in 1h 14m 36s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-openshift/48412f2/ : SUCCESS in 2h 01m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized/bb65573/ : SUCCESS in 1h 24m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-ovn/ac61296/ : FAILURE in 1h 25m 48s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-py36/f494a17/ : SUCCESS in 1h 15m 15s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-py36/5db61ff/ : SUCCESS in 1h 32m 36s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-multinode-containerized/f25969b/ : SUCCESS in 1h 52m 57s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-centos-7/e3286f4/ : SUCCESS in 1h 25m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c8d15d643432a7624629b74ea329600e9e068079","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-14 11:59:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-openshift/f81b16a/ : FAILURE in 2h 25m 18s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-l2/87f5fc2/ : FAILURE in 1h 29m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/52f8fa0/ : TIMED_OUT in 3h 03m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-network-policy/a9d7b7e/ : SUCCESS in 1h 14m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/29588cb/ : TIMED_OUT in 3h 03m 02s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-ovn/838a765/ : FAILURE in 1h 33m 29s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-openshift-ingress/861b3e7/ : FAILURE in 1h 47m 42s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/dd1ac55/ : FAILURE in 2h 00m 15s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-multinode-ha/85defa6/ : TIMED_OUT in 2h 14m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-crio/422866e/ : FAILURE in 21m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"696c3aab437893bc89a2f0023d3d287010f7a4af","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-15 13:47:39.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"fe55496857e06a91a632f2d1ac06d3eb7e32e5ab","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-15 13:47:48.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"8d579ca6ed73ee450f1819d67f012b53760a273c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 15:39:59.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/5/check/openstack-tox-docs/8e87285/ : FAILURE in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/5/check/openstack-tox-pep8/751e65a/ : SUCCESS in 4m 44s\n- kuryr-kubernetes-tempest http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest/94e4c03/ : SUCCESS in 1h 19m 11s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-openshift/cb43f1e/ : FAILURE in 1h 49m 03s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized/afbe8ae/ : SUCCESS in 1h 39m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-ovn/375b924/ : FAILURE in 1h 42m 52s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-py36/100305b/ : SUCCESS in 1h 28m 57s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-py36/dd2c240/ : FAILURE in 43m 04s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-multinode-containerized/ea9c98e/ : SUCCESS in 1h 31m 16s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-centos-7/fa05b08/ : SUCCESS in 1h 38m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"871327c71652e378764cbe6a7e5bc69b7c33caed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 16:59:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-openshift/d37e14f/ : FAILURE in 1h 31m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-l2/ab86498/ : FAILURE in 1h 39m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/a5d719e/ : TIMED_OUT in 3h 05m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-network-policy/5eca727/ : SUCCESS in 1h 00m 07s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/3351d09/ : FAILURE in 3h 04m 09s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-ovn/907b1fd/ : FAILURE in 1h 32m 54s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-openshift-ingress/e639ecf/ : FAILURE in 1h 07m 40s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/df2a986/ : FAILURE in 1h 56m 36s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-multinode-ha/c25563d/ : TIMED_OUT in 2h 16m 04s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-crio/c927da2/ : FAILURE in 25m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d0d6c4052e05dd425846e75d557f0e2630096f19","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-20 14:06:09.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"8cb343de9d1e0b5eec68f2cc38fd59c96162a359","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-20 14:06:19.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"9a9809d9290d3698ed348293c2f93e0342c00735","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 16:08:57.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/5/check/openstack-tox-docs/e05af2d/html/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/5/check/openstack-tox-pep8/e928d9d/ : SUCCESS in 4m 25s\n- kuryr-kubernetes-tempest http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest/0dde826/ : SUCCESS in 1h 24m 32s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-openshift/7b6239c/ : FAILURE in 1h 58m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized/3e752fd/ : FAILURE in 29m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-ovn/bf8bd4a/ : FAILURE in 1h 48m 22s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-py36/cf540ff/ : SUCCESS in 1h 19m 39s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-py36/98c4616/ : FAILURE in 48m 17s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-multinode-containerized/440a5da/ : FAILURE in 49m 31s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-centos-7/f12f255/ : SUCCESS in 1h 49m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f80b5a4087e2596de9f3842ecbc5139de568d0ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 17:15:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-openshift/955d8c7/ : FAILURE in 2h 14m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-l2/95dfd04/ : FAILURE in 1h 32m 55s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/0196b67/ : TIMED_OUT in 3h 03m 53s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-network-policy/e9e8d50/ : SUCCESS in 58m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/5332f35/ : SUCCESS in 2h 54m 53s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-ovn/e0c06ab/ : FAILURE in 1h 55m 38s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-openshift-ingress/2f5dcf4/ : FAILURE in 1h 02m 41s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/693a4ee/ : FAILURE in 1h 29m 14s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-multinode-ha/3c2f873/ : TIMED_OUT in 2h 14m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-crio/4ba85af/ : FAILURE in 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a3016c734cf7b645b076a908373a1994168a9699","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-21 10:43:16.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4ec37f879f0c862108986be3b6469a76ef499fdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 13:04:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/5/check/openstack-tox-docs/77568e0/html/ : SUCCESS in 5m 53s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/5/check/openstack-tox-pep8/c186468/ : SUCCESS in 5m 23s\n- kuryr-kubernetes-tempest http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest/232c470/ : SUCCESS in 1h 29m 19s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-openshift/5ed025d/ : FAILURE in 2h 15m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized/13e92ac/ : SUCCESS in 2h 04m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-ovn/2c9c7f9/ : FAILURE in 1h 43m 06s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-py36/4b58e1d/ : SUCCESS in 1h 35m 20s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-containerized-py36/9a849c0/ : SUCCESS in 1h 38m 06s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-multinode-containerized/bb21380/ : SUCCESS in 1h 45m 50s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/92/626892/5/check/kuryr-kubernetes-tempest-centos-7/99c3b1c/ : SUCCESS in 1h 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"80a10c88f825b68236ca522d17e61ad54878fc7d","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-26 10:38:46.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"7cb4b0242bd261c3b4627d857e58c7144db49d7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 13:47:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-openshift/ae8b409/ : FAILURE in 2h 07m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-l2/580bd9b/ : FAILURE in 1h 46m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/82dc69e/ : TIMED_OUT in 3h 03m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-network-policy/9943d4a/ : SUCCESS in 58m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/678d9e2/ : SUCCESS in 2h 50m 10s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-ovn/a002bfe/ : FAILURE in 1h 26m 15s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-openshift-ingress/5e7ae8a/ : FAILURE in 1h 12m 18s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/d54e1da/ : SUCCESS in 1h 38m 05s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-multinode-ha/9d0e824/ : TIMED_OUT in 2h 14m 46s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/92/626892/5/experimental/kuryr-kubernetes-tempest-containerized-crio/4233e9d/ : FAILURE in 27m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"edefc95315ca0f37a53099c32f7415a6b02f0c06","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-05-24 10:00:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a4bfdb6d7f9c1e89d863d1ac2b41d61e6794a1c6","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-05-24 10:01:02.000000000","message":"Patch Set 6:\n\ncheck experimental","accounts_in_message":[],"_revision_number":6},{"id":"ecedfa0df13e583e03728f474a1e20ccd61a97cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 12:38:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/6/check/openstack-tox-docs/49b8999/html/ : SUCCESS in 6m 25s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/6/check/openstack-tox-pep8/31fefef/ : SUCCESS in 5m 43s\n- kuryr-kubernetes-tempest http://logs.openstack.org/92/626892/6/check/kuryr-kubernetes-tempest/06ce0de/ : SUCCESS in 1h 26m 25s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/92/626892/6/check/kuryr-kubernetes-tempest-openshift/40555dd/ : FAILURE in 2h 28m 45s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/92/626892/6/check/kuryr-kubernetes-tempest-containerized/11b307b/ : SUCCESS in 1h 39m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/92/626892/6/check/kuryr-kubernetes-tempest-containerized-ovn/5d1b3ac/ : FAILURE in 2h 11m 22s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/92/626892/6/check/kuryr-kubernetes-tempest-py36/4cb2b55/ : SUCCESS in 1h 22m 30s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/92/626892/6/check/kuryr-kubernetes-tempest-containerized-py36/4f35b4d/ : SUCCESS in 1h 29m 43s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/92/626892/6/check/kuryr-kubernetes-tempest-multinode-containerized/055be06/ : SUCCESS in 1h 40m 53s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/92/626892/6/check/kuryr-kubernetes-tempest-centos-7/b4435fc/ : SUCCESS in 1h 39m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b2808e6b5fc2156e6aa6d077b170657091f0a2f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 13:06:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/92/626892/6/experimental/kuryr-kubernetes-tempest-containerized-openshift/870f429/ : FAILURE in 1h 56m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/92/626892/6/experimental/kuryr-kubernetes-tempest-containerized-l2/1800ce3/ : FAILURE in 1h 29m 59s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/92/626892/6/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/4817444/ : TIMED_OUT in 3h 03m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/92/626892/6/experimental/kuryr-kubernetes-tempest-containerized-network-policy/9a5da52/ : SUCCESS in 54m 11s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/92/626892/6/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/bf80b3f/ : SUCCESS in 2h 46m 15s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/92/626892/6/experimental/kuryr-kubernetes-tempest-ovn/f43988c/ : FAILURE in 1h 59m 03s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/92/626892/6/experimental/kuryr-kubernetes-tempest-openshift-ingress/cbbc4a6/ : FAILURE in 2h 35m 25s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/92/626892/6/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/cd5806a/ : FAILURE in 1h 50m 12s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/92/626892/6/experimental/kuryr-kubernetes-tempest-multinode-ha/5c1cc4d/ : TIMED_OUT in 2h 15m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/92/626892/6/experimental/kuryr-kubernetes-tempest-containerized-crio/fa22ed1/ : FAILURE in 27m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"df0d1e6ba562999fc99cecfa5e36b1187674aab8","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-05-24 14:55:38.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"e627cb505aa84d59d4ef8a9a21f10ac5ef483d67","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-05-28 09:55:36.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nI\u0027ll fix this…","accounts_in_message":[],"_revision_number":6},{"id":"5da1caa98786f28d22763898e65f7a35969d9a12","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-05-28 09:55:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2fc256025ee1e5b57c15ac6cdeab9a996ca67487","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-05-28 09:56:08.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"e941ceb286a4b000fbe4442de48a3151aa42bc50","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-05-28 10:02:01.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1577c4b8f5ab171a2ba124d01966ffc75f67e5f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 12:19:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/7/check/openstack-tox-docs/a2de83c/html/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/7/check/openstack-tox-pep8/9cf1856/ : SUCCESS in 4m 34s\n- kuryr-kubernetes-tempest http://logs.openstack.org/92/626892/7/check/kuryr-kubernetes-tempest/5a8d4dd/ : SUCCESS in 1h 30m 46s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/92/626892/7/check/kuryr-kubernetes-tempest-openshift/1e9efcd/ : FAILURE in 1h 43m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/92/626892/7/check/kuryr-kubernetes-tempest-containerized/6191328/ : SUCCESS in 1h 32m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/92/626892/7/check/kuryr-kubernetes-tempest-containerized-ovn/1134f5a/ : FAILURE in 1h 33m 16s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/92/626892/7/check/kuryr-kubernetes-tempest-py36/35715a2/ : SUCCESS in 1h 32m 24s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/92/626892/7/check/kuryr-kubernetes-tempest-containerized-py36/51b33ff/ : SUCCESS in 1h 38m 50s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/92/626892/7/check/kuryr-kubernetes-tempest-multinode-containerized/add3320/ : SUCCESS in 1h 56m 36s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/92/626892/7/check/kuryr-kubernetes-tempest-centos-7/4b5edd5/ : SUCCESS in 2h 09m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"104a2ad6a36c1780fcc9947e7ab09a48edf3ee96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 12:20:04.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f8fbe30b447334eb79544ea85a65edd9e2631048","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-05-28 12:24:54.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"975a45ef0f1b79afc1a0d175cfd71348266fb15d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 13:48:57.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/92/626892/7/gate/openstack-tox-docs/10450ab/html/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/92/626892/7/gate/openstack-tox-pep8/d95cea5/ : SUCCESS in 4m 07s\n- kuryr-kubernetes-tempest http://logs.openstack.org/92/626892/7/gate/kuryr-kubernetes-tempest/6f2b21e/ : SUCCESS in 1h 23m 50s\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/92/626892/7/gate/kuryr-kubernetes-tempest-py36/50a2f3e/ : SUCCESS in 1h 17m 52s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/92/626892/7/gate/kuryr-kubernetes-tempest-containerized-py36/ea105bd/ : SUCCESS in 1h 22m 06s","accounts_in_message":[],"_revision_number":7},{"id":"34410b7de0980c732750557cfe4b9f496dff569b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 13:48:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6c4ac2bebd235264f0787006d3478032dee450ea","revisions":{"c86c9c9912a1d880525a785ad60fffb30b28a630":{"kind":"REWORK","_number":1,"created":"2018-12-21 13:21:41.000000000","uploader":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"ref":"refs/changes/92/626892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/92/626892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/1"}}},"commit":{"parents":[{"commit":"bec3f7fe63f527c717501de5f648f3d6666f4b8f","subject":"Multi worker tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bec3f7fe63f527c717501de5f648f3d6666f4b8f"}]}],"author":{"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","date":"2018-12-21 13:20:09.000000000","tz":60},"committer":{"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","date":"2018-12-21 13:20:09.000000000","tz":60},"subject":"detect failed curl when streamed from pod","message":"detect failed curl when streamed from pod\n\nWithout this code, the error was a very unspecific failure to convert\nthe status code to integer. In this way, we\u0027ll know if curl actually got\nto run or there was some issue in kubelet streaming from the pod.\n\nChange-Id: Ia2be9260c31fcde0da508eb3d7021d69b99f27d2\nSigned-off-by: Antoni Segura Puimedon \u003ccelebdor@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c86c9c9912a1d880525a785ad60fffb30b28a630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c86c9c9912a1d880525a785ad60fffb30b28a630"}]},"branch":"refs/heads/master"},"3bf3ad6a76c8c81365614abd530ea4adb9842ebb":{"kind":"REWORK","_number":2,"created":"2018-12-24 08:42:55.000000000","uploader":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"ref":"refs/changes/92/626892/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/92/626892/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/2"}}},"commit":{"parents":[{"commit":"bec3f7fe63f527c717501de5f648f3d6666f4b8f","subject":"Multi worker tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bec3f7fe63f527c717501de5f648f3d6666f4b8f"}]}],"author":{"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","date":"2018-12-21 13:20:09.000000000","tz":60},"committer":{"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","date":"2018-12-24 08:42:46.000000000","tz":60},"subject":"detect failed curl when streamed from pod","message":"detect failed curl when streamed from pod\n\nWithout this code, the error was a very unspecific failure to convert\nthe status code to integer. In this way, we\u0027ll know if curl actually got\nto run or there was some issue in kubelet streaming from the pod.\n\nChange-Id: Ia2be9260c31fcde0da508eb3d7021d69b99f27d2\nSigned-off-by: Antoni Segura Puimedon \u003ccelebdor@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3bf3ad6a76c8c81365614abd530ea4adb9842ebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3bf3ad6a76c8c81365614abd530ea4adb9842ebb"}]},"branch":"refs/heads/master"},"3813a886e28a98da380456351897cdd1f65aa43f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-01-07 10:16:57.000000000","uploader":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"ref":"refs/changes/92/626892/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/92/626892/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/3"}}},"commit":{"parents":[{"commit":"7f84979019b4cfc44927bd494c4c37985e6aa209","subject":"Merge \"Making sure all the pods reside on the same node for port_pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/7f84979019b4cfc44927bd494c4c37985e6aa209"}]}],"author":{"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","date":"2018-12-21 13:20:09.000000000","tz":60},"committer":{"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","date":"2019-01-07 10:16:57.000000000","tz":0},"subject":"detect failed curl when streamed from pod","message":"detect failed curl when streamed from pod\n\nWithout this code, the error was a very unspecific failure to convert\nthe status code to integer. In this way, we\u0027ll know if curl actually got\nto run or there was some issue in kubelet streaming from the pod.\n\nChange-Id: Ia2be9260c31fcde0da508eb3d7021d69b99f27d2\nSigned-off-by: Antoni Segura Puimedon \u003ccelebdor@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3813a886e28a98da380456351897cdd1f65aa43f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/3813a886e28a98da380456351897cdd1f65aa43f"}]},"branch":"refs/heads/master"},"66c202c90be74e3df318d22a976a09537de3a38c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-03-14 08:10:47.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/92/626892/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/92/626892/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/4"}}},"commit":{"parents":[{"commit":"0d51e9905bf9d9af62d6f536b7c2156956c59699","subject":"Test CRD podSelector update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0d51e9905bf9d9af62d6f536b7c2156956c59699"}]}],"author":{"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","date":"2018-12-21 13:20:09.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-03-14 08:10:47.000000000","tz":0},"subject":"detect failed curl when streamed from pod","message":"detect failed curl when streamed from pod\n\nWithout this code, the error was a very unspecific failure to convert\nthe status code to integer. In this way, we\u0027ll know if curl actually got\nto run or there was some issue in kubelet streaming from the pod.\n\nChange-Id: Ia2be9260c31fcde0da508eb3d7021d69b99f27d2\nSigned-off-by: Antoni Segura Puimedon \u003ccelebdor@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/66c202c90be74e3df318d22a976a09537de3a38c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/66c202c90be74e3df318d22a976a09537de3a38c"}]},"branch":"refs/heads/master"},"4ba301a3cb83baebfd1148c98aacd67467c4e5c9":{"kind":"REWORK","_number":5,"created":"2019-03-31 13:20:25.000000000","uploader":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"ref":"refs/changes/92/626892/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/92/626892/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/5"}}},"commit":{"parents":[{"commit":"0d51e9905bf9d9af62d6f536b7c2156956c59699","subject":"Test CRD podSelector update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0d51e9905bf9d9af62d6f536b7c2156956c59699"}]}],"author":{"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","date":"2018-12-21 13:20:09.000000000","tz":60},"committer":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-03-31 13:18:56.000000000","tz":180},"subject":"detect failed curl when streamed from pod","message":"detect failed curl when streamed from pod\n\nWithout this code, the error was a very unspecific failure to convert\nthe status code to integer. In this way, we\u0027ll know if curl actually got\nto run or there was some issue in kubelet streaming from the pod.\n\nChange-Id: Ia2be9260c31fcde0da508eb3d7021d69b99f27d2\nSigned-off-by: Antoni Segura Puimedon \u003ccelebdor@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4ba301a3cb83baebfd1148c98aacd67467c4e5c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4ba301a3cb83baebfd1148c98aacd67467c4e5c9"}]},"branch":"refs/heads/master"},"8395865c8c98a212f3c3460e92a98ff12fd8853a":{"kind":"REWORK","_number":6,"created":"2019-05-24 10:00:06.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/92/626892/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/92/626892/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/6"}}},"commit":{"parents":[{"commit":"c9ef83d0fb6335a8728a8c2fe796d2f30168295b","subject":"Create/delete network policy in non-default namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c9ef83d0fb6335a8728a8c2fe796d2f30168295b"}]}],"author":{"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","date":"2018-12-21 13:20:09.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-05-24 09:58:18.000000000","tz":120},"subject":"detect failed curl when streamed from pod","message":"detect failed curl when streamed from pod\n\nWithout this code, the error was a very unspecific failure to convert\nthe status code to integer. In this way, we\u0027ll know if curl actually got\nto run or there was some issue in kubelet streaming from the pod.\n\nChange-Id: Ia2be9260c31fcde0da508eb3d7021d69b99f27d2\nSigned-off-by: Antoni Segura Puimedon \u003ccelebdor@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8395865c8c98a212f3c3460e92a98ff12fd8853a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8395865c8c98a212f3c3460e92a98ff12fd8853a"}]},"branch":"refs/heads/master"},"6c4ac2bebd235264f0787006d3478032dee450ea":{"kind":"REWORK","_number":7,"created":"2019-05-28 09:55:47.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/92/626892/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/92/626892/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/92/626892/7"}}},"commit":{"parents":[{"commit":"c9ef83d0fb6335a8728a8c2fe796d2f30168295b","subject":"Create/delete network policy in non-default namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c9ef83d0fb6335a8728a8c2fe796d2f30168295b"}]}],"author":{"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","date":"2018-12-21 13:20:09.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-05-28 09:55:18.000000000","tz":120},"subject":"detect failed curl when streamed from pod","message":"detect failed curl when streamed from pod\n\nWithout this code, the error was a very unspecific failure to convert\nthe status code to integer. In this way, we\u0027ll know if curl actually got\nto run or there was some issue in kubelet streaming from the pod.\n\nChange-Id: Ia2be9260c31fcde0da508eb3d7021d69b99f27d2\nSigned-off-by: Antoni Segura Puimedon \u003ccelebdor@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6c4ac2bebd235264f0787006d3478032dee450ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6c4ac2bebd235264f0787006d3478032dee450ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
