)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~652276","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I8a0a4b3a8ef18072eef0db05eb98eb1f7ca57553","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"create_delete_np_non_default_ns","hashtags":[],"change_id":"I8a0a4b3a8ef18072eef0db05eb98eb1f7ca57553","subject":"Create/delete network policy in non-default namespace","status":"MERGED","created":"2019-04-14 12:14:54.000000000","updated":"2019-05-03 19:56:01.000000000","submitted":"2019-05-03 19:56:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652276-1556913361542-76a6aaaf","meta_rev_id":"0acb094e0d828333af4910d130814a188ae28d41","_number":652276,"virtual_id_number":652276,"owner":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"date":"2019-04-28 13:44:44.000000000","_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":2,"date":"2019-05-03 19:56:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2019-05-03 17:56:13.000000000","_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-29 12:34:20.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"all":[{"value":0,"date":"2019-04-30 07:49:38.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2019-05-03 17:56:13.000000000","_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-29 12:34:20.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-04-30 07:49:38.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-05-03 17:56:13.000000000","updated_by":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"reviewer":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"state":"REVIEWER"},{"updated":"2019-05-03 19:56:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f40941245ca6ea7f6c841b577d38ad7b326d4a50","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-14 12:14:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50083a4fce47f75ce2339b24e596a0d420855c00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-14 14:29:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/76/652276/1/check/openstack-tox-docs/121a601/ : FAILURE in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/76/652276/1/check/openstack-tox-pep8/74a0b3a/ : SUCCESS in 4m 06s\n- kuryr-kubernetes-tempest http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest/db3aad8/ : SUCCESS in 1h 16m 40s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-openshift/6a5dae4/ : FAILURE in 1h 58m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-containerized/0ec1086/ : SUCCESS in 1h 37m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-containerized-ovn/318163c/ : FAILURE in 1h 49m 32s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-py36/af67afc/ : SUCCESS in 1h 22m 16s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-containerized-py36/5ad20d8/ : SUCCESS in 2h 06m 02s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-multinode-containerized/f12de6f/ : SUCCESS in 1h 39m 15s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-centos-7/19958e4/ : FAILURE in 2h 08m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e70e8c413deec111fd20697f4caaa0c5657e48f","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-04-15 07:01:09.000000000","message":"Patch Set 1:\n\nwhat is the difference between this one and https://review.openstack.org/#/c/648859/?","accounts_in_message":[],"_revision_number":1},{"id":"f27a297490247a9677beffa7e5742c7b9cca3d75","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-15 07:36:25.000000000","message":"Patch Set 1:\n\n\u003e what is the difference between this one and https://review.openstack.org/#/c/648859/?\n\nIn this one I remove network policy and in the previous one I remove namespace","accounts_in_message":[],"_revision_number":1},{"id":"982745ece6b0ffa1f660886a907047b89d2ed984","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-04-16 07:27:17.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e \u003e what is the difference between this one and https://review.openstack.org/#/c/648859/?\n \u003e \n \u003e In this one I remove network policy and in the previous one I\n \u003e remove namespace\n\nOk, seems like a bit of duplication as it should not matter if the NP is created on the default or any other namespace, but it should not harm anyway","accounts_in_message":[],"_revision_number":1},{"id":"1fc10a5266f772822d6f9cae82e472f292b0cd4b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-04-16 07:27:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dcdbaad67e411312bddca51de476831f80aa7fed","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-04-16 07:27:30.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"71274274b0af179daaf5a101e9517ee182337790","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 09:50:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/76/652276/1/check/openstack-tox-docs/97f868e/ : FAILURE in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/76/652276/1/check/openstack-tox-pep8/dc81665/ : SUCCESS in 4m 47s\n- kuryr-kubernetes-tempest http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest/5e84e17/ : SUCCESS in 1h 31m 31s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-openshift/f264b08/ : FAILURE in 2h 02m 32s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-containerized/c22d355/ : SUCCESS in 1h 24m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-containerized-ovn/1ef5933/ : FAILURE in 1h 57m 48s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-py36/7f6c609/ : SUCCESS in 1h 32m 29s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-containerized-py36/2ae1975/ : SUCCESS in 1h 42m 46s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-multinode-containerized/58641dd/ : SUCCESS in 1h 37m 27s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/76/652276/1/check/kuryr-kubernetes-tempest-centos-7/61cf654/ : SUCCESS in 1h 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4d2dbc059e58ce8cb072502c7ad746ca6326814","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 10:41:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/76/652276/1/experimental/kuryr-kubernetes-tempest-containerized-openshift/013f237/ : SUCCESS in 1h 55m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/76/652276/1/experimental/kuryr-kubernetes-tempest-containerized-l2/7ec5bc0/ : FAILURE in 2h 02m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/76/652276/1/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/1f69685/ : TIMED_OUT in 3h 04m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/76/652276/1/experimental/kuryr-kubernetes-tempest-containerized-network-policy/71aceaf/ : FAILURE in 1h 01m 28s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/76/652276/1/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/a481ea5/ : SUCCESS in 2h 57m 16s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/76/652276/1/experimental/kuryr-kubernetes-tempest-ovn/6567d40/ : FAILURE in 1h 40m 53s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/76/652276/1/experimental/kuryr-kubernetes-tempest-openshift-ingress/b351c9f/ : FAILURE in 1h 03m 43s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/76/652276/1/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/3e7a7f6/ : FAILURE in 2h 04m 03s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/76/652276/1/experimental/kuryr-kubernetes-tempest-multinode-ha/c1de206/ : TIMED_OUT in 2h 15m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/76/652276/1/experimental/kuryr-kubernetes-tempest-containerized-crio/c31f7a9/ : FAILURE in 27m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65aa58b938a61f2de7d940f4d6efefa45109097a","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-16 12:46:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6ad36f623e09c0769ae6940c639fb515a0442044","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-16 12:47:03.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"d57949737c83226f7bf8abfb90a0b649123d382f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 15:55:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/76/652276/2/experimental/kuryr-kubernetes-tempest-containerized-openshift/c2dc6dc/ : FAILURE in 1h 46m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/76/652276/2/experimental/kuryr-kubernetes-tempest-containerized-l2/79043ce/ : FAILURE in 1h 50m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/76/652276/2/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/e38d556/ : TIMED_OUT in 3h 04m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/76/652276/2/experimental/kuryr-kubernetes-tempest-containerized-network-policy/cf784bd/ : FAILURE in 1h 05m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/76/652276/2/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/3b34452/ : FAILURE in 3h 03m 36s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/76/652276/2/experimental/kuryr-kubernetes-tempest-ovn/120eef9/ : FAILURE in 1h 46m 58s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/76/652276/2/experimental/kuryr-kubernetes-tempest-openshift-ingress/4df9c3f/ : FAILURE in 57m 53s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/76/652276/2/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/059ac6e/ : FAILURE in 2h 04m 43s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/76/652276/2/experimental/kuryr-kubernetes-tempest-multinode-ha/a37cfba/ : TIMED_OUT in 2h 15m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/76/652276/2/experimental/kuryr-kubernetes-tempest-containerized-crio/412018e/ : FAILURE in 28m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72b7e8181ab468286016a74ae58ddd22bc0f793d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 15:57:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/76/652276/2/check/openstack-tox-docs/806047f/html/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/76/652276/2/check/openstack-tox-pep8/1889a07/ : SUCCESS in 4m 33s\n- kuryr-kubernetes-tempest http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest/29f6715/ : SUCCESS in 1h 33m 48s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-openshift/29fc6a6/ : TIMED_OUT in 3h 05m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-containerized/ec10f87/ : SUCCESS in 1h 21m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-containerized-ovn/e429971/ : FAILURE in 2h 08m 59s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-py36/ca5544e/ : SUCCESS in 1h 32m 25s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-containerized-py36/27f7f22/ : FAILURE in 1h 50m 10s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-multinode-containerized/9f4a746/ : FAILURE in 1h 49m 04s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-centos-7/c9142ff/ : FAILURE in 55m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9224d9148d63dab207c3f544a8430e0e6695f5be","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-17 08:42:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"44bad7eacc1628624bf07ccb103f60ad45b241a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 10:50:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/76/652276/2/check/openstack-tox-docs/1b6d348/html/ : SUCCESS in 6m 15s\n- openstack-tox-pep8 http://logs.openstack.org/76/652276/2/check/openstack-tox-pep8/bd506e6/ : SUCCESS in 7m 22s\n- kuryr-kubernetes-tempest http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest/20337ec/ : SUCCESS in 1h 22m 53s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-openshift/232a61b/ : FAILURE in 1h 50m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-containerized/33aa415/ : SUCCESS in 1h 31m 51s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-containerized-ovn/9b8f37e/ : FAILURE in 1h 46m 25s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-py36/5b4b554/ : SUCCESS in 1h 34m 43s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-containerized-py36/fc80d59/ : SUCCESS in 1h 41m 45s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-multinode-containerized/174a18a/ : SUCCESS in 1h 57m 48s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/76/652276/2/check/kuryr-kubernetes-tempest-centos-7/275f9c9/ : SUCCESS in 2h 01m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a77ce511711c5b3db589427b28ab0fa62154bb29","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-18 12:26:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"38bae3e944332aab31e083a5acc83f81d285cd44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 15:32:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/76/652276/3/check/openstack-tox-docs/32858bd/html/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/76/652276/3/check/openstack-tox-pep8/2194eed/ : SUCCESS in 4m 54s\n- kuryr-kubernetes-tempest http://logs.openstack.org/76/652276/3/check/kuryr-kubernetes-tempest/103468b/ : SUCCESS in 1h 36m 32s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/76/652276/3/check/kuryr-kubernetes-tempest-openshift/3cd63b0/ : FAILURE in 2h 46m 37s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/76/652276/3/check/kuryr-kubernetes-tempest-containerized/a842d17/ : SUCCESS in 1h 59m 16s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/76/652276/3/check/kuryr-kubernetes-tempest-containerized-ovn/9058db4/ : FAILURE in 1h 53m 10s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/76/652276/3/check/kuryr-kubernetes-tempest-py36/a53996f/ : SUCCESS in 1h 27m 31s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/76/652276/3/check/kuryr-kubernetes-tempest-containerized-py36/99f8b41/ : SUCCESS in 1h 41m 03s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/76/652276/3/check/kuryr-kubernetes-tempest-multinode-containerized/1c6838e/ : SUCCESS in 1h 45m 58s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/76/652276/3/check/kuryr-kubernetes-tempest-centos-7/d7025de/ : SUCCESS in 1h 27m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"502fd5148e7b3d474a6cf29bd7c75aa9c1aed280","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-20 14:05:33.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"1e3b782aadb74f08f5136ecc43141a9928ac7afe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 17:10:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/76/652276/3/experimental/kuryr-kubernetes-tempest-containerized-openshift/96eb96c/ : SUCCESS in 1h 35m 26s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/76/652276/3/experimental/kuryr-kubernetes-tempest-containerized-l2/4251d15/ : FAILURE in 1h 45m 52s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/76/652276/3/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/a1fd659/ : TIMED_OUT in 3h 03m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/76/652276/3/experimental/kuryr-kubernetes-tempest-containerized-network-policy/2ec9294/ : SUCCESS in 1h 01m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/76/652276/3/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/a447df6/ : SUCCESS in 2h 40m 26s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/76/652276/3/experimental/kuryr-kubernetes-tempest-ovn/13248b0/ : FAILURE in 1h 42m 53s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/76/652276/3/experimental/kuryr-kubernetes-tempest-openshift-ingress/b89cbb2/ : TIMED_OUT in 3h 04m 04s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/76/652276/3/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/f941016/ : FAILURE in 2h 08m 02s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/76/652276/3/experimental/kuryr-kubernetes-tempest-multinode-ha/a04fb00/ : TIMED_OUT in 2h 14m 56s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/76/652276/3/experimental/kuryr-kubernetes-tempest-containerized-crio/27e79f3/ : FAILURE in 25m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d4a64369aa667e24dba7bf2f423cc63ea704363b","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-28 13:43:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"369e394a815ca6383e1eb756f3a74e55661c22a1","author":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"date":"2019-04-28 13:44:44.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"a1dfb20728945a8defc1aebcb6cd54a9375d2630","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 15:51:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/76/652276/4/check/openstack-tox-docs/c29b5ea/html/ : SUCCESS in 5m 51s\n- openstack-tox-pep8 http://logs.openstack.org/76/652276/4/check/openstack-tox-pep8/7f585d1/ : SUCCESS in 4m 16s\n- kuryr-kubernetes-tempest http://logs.openstack.org/76/652276/4/check/kuryr-kubernetes-tempest/7afc828/ : SUCCESS in 1h 21m 06s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/76/652276/4/check/kuryr-kubernetes-tempest-openshift/b944c29/ : FAILURE in 1h 56m 31s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/76/652276/4/check/kuryr-kubernetes-tempest-containerized/cbe5adb/ : SUCCESS in 1h 27m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/76/652276/4/check/kuryr-kubernetes-tempest-containerized-ovn/3bd7256/ : FAILURE in 2h 03m 37s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/76/652276/4/check/kuryr-kubernetes-tempest-py36/334da4a/ : SUCCESS in 1h 33m 29s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/76/652276/4/check/kuryr-kubernetes-tempest-containerized-py36/c4065f5/ : SUCCESS in 1h 50m 48s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/76/652276/4/check/kuryr-kubernetes-tempest-multinode-containerized/00a0e02/ : SUCCESS in 1h 36m 37s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/76/652276/4/check/kuryr-kubernetes-tempest-centos-7/7ad23dc/ : SUCCESS in 1h 40m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"789bda433941442b6ef172472b31a314130a036b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 16:49:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/76/652276/4/experimental/kuryr-kubernetes-tempest-containerized-openshift/893a59e/ : FAILURE in 2h 07m 35s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/76/652276/4/experimental/kuryr-kubernetes-tempest-containerized-l2/fd4c652/ : FAILURE in 1h 57m 05s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/76/652276/4/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/0e62131/ : TIMED_OUT in 3h 03m 24s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/76/652276/4/experimental/kuryr-kubernetes-tempest-containerized-network-policy/0c04339/ : SUCCESS in 46m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/76/652276/4/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/35360e3/ : FAILURE in 2h 46m 56s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/76/652276/4/experimental/kuryr-kubernetes-tempest-ovn/8839a17/ : FAILURE in 1h 54m 03s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/76/652276/4/experimental/kuryr-kubernetes-tempest-openshift-ingress/5034543/ : TIMED_OUT in 3h 03m 28s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/76/652276/4/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/7f8a4b4/ : SUCCESS in 1h 42m 34s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/76/652276/4/experimental/kuryr-kubernetes-tempest-multinode-ha/1bb0309/ : SUCCESS in 2h 10m 14s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/76/652276/4/experimental/kuryr-kubernetes-tempest-containerized-crio/fe7ace9/ : FAILURE in 26m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7015e1b14d231d14f8bd24c6d19a15dc2c3a9bf4","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-04-29 12:34:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1994bc481d1860579e177e9a9861acddd6e1d45f","author":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"date":"2019-05-03 17:56:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0aa6c113f3922ff2c45ab4f7bc826d2d3280dcca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 17:57:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2590ad5240d4005a715ec430760dcb87aa6218d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 19:56:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4153a4fa60f2d95e6b9ef0599477ca43bd5c8d45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 19:56:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/76/652276/4/gate/openstack-tox-docs/c013cb2/html/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/76/652276/4/gate/openstack-tox-pep8/797eead/ : SUCCESS in 4m 31s\n- kuryr-kubernetes-tempest http://logs.openstack.org/76/652276/4/gate/kuryr-kubernetes-tempest/c7d095d/ : SUCCESS in 1h 30m 57s\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/76/652276/4/gate/kuryr-kubernetes-tempest-py36/809fd7c/ : SUCCESS in 1h 29m 51s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/76/652276/4/gate/kuryr-kubernetes-tempest-containerized-py36/369bb6f/ : SUCCESS in 1h 48m 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c9ef83d0fb6335a8728a8c2fe796d2f30168295b","revisions":{"da5a3ec10f4f880629fd3a504f6b25558beac0ad":{"kind":"REWORK","_number":1,"created":"2019-04-14 12:14:54.000000000","uploader":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"ref":"refs/changes/76/652276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/76/652276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/1"}}},"commit":{"parents":[{"commit":"566419095449702aac9f9740f8ed54378b7f096a","subject":"Delete namespace with network policy on it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/566419095449702aac9f9740f8ed54378b7f096a"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-04-14 12:11:31.000000000","tz":180},"committer":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-04-14 12:11:31.000000000","tz":180},"subject":"Create/delete network policy in non-default namespace","message":"Create/delete network policy in non-default namespace\n\nCreate network policy on non-default namespace and make sure that\nCRD and SG were created for that policy\nRemove network policy in non-default namespace and make sure\nthe network policy CRD and SG were deleted\n\nChange-Id: I8a0a4b3a8ef18072eef0db05eb98eb1f7ca57553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/da5a3ec10f4f880629fd3a504f6b25558beac0ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/da5a3ec10f4f880629fd3a504f6b25558beac0ad"}]},"branch":"refs/heads/master"},"910e23c64414e4daf93c6753839120222c686932":{"kind":"REWORK","_number":2,"created":"2019-04-16 12:46:06.000000000","uploader":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"ref":"refs/changes/76/652276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/76/652276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/2"}}},"commit":{"parents":[{"commit":"566419095449702aac9f9740f8ed54378b7f096a","subject":"Delete namespace with network policy on it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/566419095449702aac9f9740f8ed54378b7f096a"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-04-14 12:11:31.000000000","tz":180},"committer":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-04-16 12:45:53.000000000","tz":180},"subject":"Create/delete network policy in non-default namespace","message":"Create/delete network policy in non-default namespace\n\nCreate network policy on non-default namespace and make sure that\nCRD and SG were created for that policy\nRemove network policy in non-default namespace and make sure\nthe network policy CRD and SG were deleted\n\nChange-Id: I8a0a4b3a8ef18072eef0db05eb98eb1f7ca57553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/910e23c64414e4daf93c6753839120222c686932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/910e23c64414e4daf93c6753839120222c686932"}]},"branch":"refs/heads/master"},"339e26cd0fe625d9ba8e00c569de67e210a2dac2":{"kind":"REWORK","_number":3,"created":"2019-04-18 12:26:12.000000000","uploader":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"ref":"refs/changes/76/652276/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/76/652276/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/3"}}},"commit":{"parents":[{"commit":"566419095449702aac9f9740f8ed54378b7f096a","subject":"Delete namespace with network policy on it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/566419095449702aac9f9740f8ed54378b7f096a"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-04-14 12:11:31.000000000","tz":180},"committer":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-04-18 12:25:57.000000000","tz":180},"subject":"Create/delete network policy in non-default namespace","message":"Create/delete network policy in non-default namespace\n\nCreate network policy on non-default namespace and make sure that\nCRD and SG were created for that policy\nRemove network policy in non-default namespace and make sure\nthe network policy CRD and SG were deleted\n\nChange-Id: I8a0a4b3a8ef18072eef0db05eb98eb1f7ca57553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/339e26cd0fe625d9ba8e00c569de67e210a2dac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/339e26cd0fe625d9ba8e00c569de67e210a2dac2"}]},"branch":"refs/heads/master"},"c9ef83d0fb6335a8728a8c2fe796d2f30168295b":{"kind":"REWORK","_number":4,"created":"2019-04-28 13:43:03.000000000","uploader":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"ref":"refs/changes/76/652276/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/76/652276/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/76/652276/4"}}},"commit":{"parents":[{"commit":"5df10697a491e00cc2e2e2bdea2f4296e21e20ac","subject":"Merge \"Delete namespace with network policy on it\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5df10697a491e00cc2e2e2bdea2f4296e21e20ac"}]}],"author":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-04-14 12:11:31.000000000","tz":180},"committer":{"name":"Genadi Chereshnya","email":"gcheresh@redhat.com","date":"2019-04-28 13:42:38.000000000","tz":180},"subject":"Create/delete network policy in non-default namespace","message":"Create/delete network policy in non-default namespace\n\nCreate network policy on non-default namespace and make sure that\nCRD and SG were created for that policy\nRemove network policy in non-default namespace and make sure\nthe network policy CRD and SG were deleted\n\nChange-Id: I8a0a4b3a8ef18072eef0db05eb98eb1f7ca57553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c9ef83d0fb6335a8728a8c2fe796d2f30168295b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c9ef83d0fb6335a8728a8c2fe796d2f30168295b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
