)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~653372","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Id0fa12c3bb240d1d8eae01db78558e7f908f80a9","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"bug/1799239","hashtags":[],"change_id":"Id0fa12c3bb240d1d8eae01db78558e7f908f80a9","subject":"Update rootfs.tar.xz for kuryr/demo","status":"MERGED","created":"2019-04-17 09:44:46.000000000","updated":"2019-04-23 23:25:49.000000000","submitted":"2019-04-23 23:25:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653372-1556061949665-dfe5b5ea","meta_rev_id":"7a503728432f1bc32de7fe74cc50fddacb3408ab","_number":653372,"virtual_id_number":653372,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-23 19:31:39.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":2,"date":"2019-04-23 23:25:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2019-04-22 09:57:09.000000000","_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-17 12:34:41.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2019-04-22 09:57:09.000000000","_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"date":"2019-04-17 11:55:06.000000000","_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 11:55:06.000000000","updated_by":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"reviewer":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"state":"REVIEWER"},{"updated":"2019-04-17 12:34:41.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-04-22 09:57:09.000000000","updated_by":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"reviewer":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"state":"REVIEWER"},{"updated":"2019-04-23 23:25:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"281514933170b356cf21c4cb78a2e00693903138","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-04-17 09:44:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccf6a26df95fd60c240993310a9bbe49f42c5143","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 11:49:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/653372/1/check/openstack-tox-docs/24e11d9/html/ : SUCCESS in 5m 22s\n- openstack-tox-pep8 http://logs.openstack.org/72/653372/1/check/openstack-tox-pep8/a476c2f/ : SUCCESS in 5m 57s\n- kuryr-kubernetes-tempest http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest/7c2de3e/ : SUCCESS in 1h 44m 26s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-openshift/f9ee593/ : FAILURE in 1h 37m 43s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-containerized/ff0dc0d/ : SUCCESS in 1h 30m 09s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-containerized-ovn/4efed5a/ : FAILURE in 1h 46m 55s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-py36/c18fbbe/ : SUCCESS in 1h 21m 49s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-containerized-py36/88128bd/ : SUCCESS in 1h 43m 54s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-multinode-containerized/fef289e/ : SUCCESS in 1h 40m 07s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-centos-7/09f4563/ : SUCCESS in 1h 31m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8456838704f2f6d272dcdaae90484abe1c9ca64a","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-04-17 12:34:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"919e2081df08d5ed27ca10599c6990a3f073780a","author":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"date":"2019-04-22 09:57:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2145722d4f6a84ca1e0807e820515fdbed514cc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 09:57:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d39c34c9d3d12da58721d139b93cb03268e6b3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:55:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/653372/1/gate/openstack-tox-docs/604ea71/html/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/72/653372/1/gate/openstack-tox-pep8/ead43ff/ : SUCCESS in 4m 41s\n- kuryr-kubernetes-tempest http://logs.openstack.org/72/653372/1/gate/kuryr-kubernetes-tempest/22cd4a2/ : FAILURE in 1h 29m 31s\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/72/653372/1/gate/kuryr-kubernetes-tempest-py36/f8aae63/ : SUCCESS in 1h 37m 37s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/72/653372/1/gate/kuryr-kubernetes-tempest-containerized-py36/84e11ca/ : SUCCESS in 1h 56m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2736678cfa620923c2d28569860231e5a1308388","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-04-23 07:37:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"53657d51f33d4df82005e2040304a7759ebdd4b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:50:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/653372/1/check/openstack-tox-docs/9760d3d/html/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/72/653372/1/check/openstack-tox-pep8/9fa25d5/ : SUCCESS in 4m 48s\n- kuryr-kubernetes-tempest http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest/158c964/ : SUCCESS in 1h 37m 12s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-openshift/e976640/ : FAILURE in 1h 49m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-containerized/3771893/ : FAILURE in 1h 55m 10s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-containerized-ovn/beb768f/ : FAILURE in 1h 59m 05s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-py36/42dc3fb/ : SUCCESS in 1h 22m 24s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-containerized-py36/2b70da3/ : SUCCESS in 1h 34m 49s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-multinode-containerized/29af29c/ : SUCCESS in 2h 02m 58s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-centos-7/30c59d4/ : FAILURE in 1h 40m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"927675fa1762c5dd60038301c5099c997c34facd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:50:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14b2383cde1e65338d1ee5533166500d72bf6cf7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:10:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/653372/1/gate/openstack-tox-docs/2facc21/html/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/72/653372/1/gate/openstack-tox-pep8/a831f18/ : SUCCESS in 4m 30s\n- kuryr-kubernetes-tempest http://logs.openstack.org/72/653372/1/gate/kuryr-kubernetes-tempest/a13b5d9/ : POST_FAILURE in 1h 30m 27s\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/72/653372/1/gate/kuryr-kubernetes-tempest-py36/38fbdbf/ : SUCCESS in 1h 19m 32s\n- kuryr-kubernetes-tempest-containerized-py36 finger://ze06.openstack.org/bae095e97f2e44889830bf0d503e16d5 : RETRY_LIMIT in 40m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b97868a6c2d130cd5e3ac9473c24449d5b1687dd","author":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"date":"2019-04-23 13:28:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"20e245e15ffd153b1a38c40e9a9c23c30202ccab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 16:32:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/72/653372/1/check/openstack-tox-docs/b425bc9/html/ : SUCCESS in 8m 19s\n- openstack-tox-pep8 http://logs.openstack.org/72/653372/1/check/openstack-tox-pep8/4c6f60d/ : SUCCESS in 6m 17s\n- kuryr-kubernetes-tempest finger://ze04.openstack.org/1461e3eb1aff43079048fe973a59044e : RETRY_LIMIT in 20m 26s\n- kuryr-kubernetes-tempest-openshift finger://ze08.openstack.org/c183d0569a984cff8215eba3e2e5145d : RETRY_LIMIT in 31m 21s (non-voting)\n- kuryr-kubernetes-tempest-containerized finger://ze10.openstack.org/12f960db073b4d078d5ca766f16154c3 : RETRY_LIMIT in 25m 13s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn finger://ze12.openstack.org/cbac136a65cd4b24a58d25a4c7fba41d : RETRY_LIMIT in 27m 53s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-py36/ed9d01c/ : FAILURE in 30m 24s\n- kuryr-kubernetes-tempest-containerized-py36 finger://ze07.openstack.org/693987a3ab96404cb8a3bfab242b2180 : RETRY_LIMIT in 30m 49s\n- kuryr-kubernetes-tempest-multinode-containerized finger://ze07.openstack.org/a59fb2eeba2f416fb92f8c4a5814a01b : RETRY_LIMIT in 21m 21s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 finger://ze07.openstack.org/ada51c20357f4e8090d2825b38c567b2 : RETRY_LIMIT in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4a920730bee95997de9a5735e8fc9f8f291b1d8","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-04-23 19:31:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f7baeefe0d7766fba054378f30effdb533e73db7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 21:45:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/653372/1/check/openstack-tox-docs/c26b14f/html/ : SUCCESS in 6m 38s\n- openstack-tox-pep8 http://logs.openstack.org/72/653372/1/check/openstack-tox-pep8/e1a3d97/ : SUCCESS in 6m 54s\n- kuryr-kubernetes-tempest http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest/df17c63/ : SUCCESS in 1h 19m 38s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-openshift/17e5a56/ : FAILURE in 2h 12m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-containerized/c82152a/ : SUCCESS in 1h 45m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-containerized-ovn/15d705f/ : FAILURE in 1h 46m 51s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-py36/edba469/ : SUCCESS in 1h 23m 16s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-containerized-py36/140d177/ : SUCCESS in 1h 35m 38s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-multinode-containerized/44adc36/ : SUCCESS in 1h 57m 51s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/72/653372/1/check/kuryr-kubernetes-tempest-centos-7/63e737b/ : SUCCESS in 1h 40m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5d2e7b9653a3a9c3165ec510c704dcce2cb4a9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 21:45:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4f7943cbe3a98b84cfbbf19b5f389ed73dcc26d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 23:25:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0627cb4b4f3370eaaa22dcb54f79ace8340a47d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 23:25:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/72/653372/1/gate/openstack-tox-docs/d369eba/html/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/72/653372/1/gate/openstack-tox-pep8/347d3ef/ : SUCCESS in 4m 15s\n- kuryr-kubernetes-tempest http://logs.openstack.org/72/653372/1/gate/kuryr-kubernetes-tempest/11361b1/ : SUCCESS in 1h 34m 35s\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/72/653372/1/gate/kuryr-kubernetes-tempest-py36/259c87e/ : SUCCESS in 1h 24m 03s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/72/653372/1/gate/kuryr-kubernetes-tempest-containerized-py36/e6cf5d7/ : SUCCESS in 1h 27m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44fa61591d545ea8eaf3f1e6be1bed9797ebca8c","revisions":{"44fa61591d545ea8eaf3f1e6be1bed9797ebca8c":{"kind":"REWORK","_number":1,"created":"2019-04-17 09:44:46.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/72/653372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/72/653372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/72/653372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/72/653372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/72/653372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/72/653372/1"}}},"commit":{"parents":[{"commit":"0d51e9905bf9d9af62d6f536b7c2156956c59699","subject":"Test CRD podSelector update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0d51e9905bf9d9af62d6f536b7c2156956c59699"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-04-17 09:38:55.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-04-17 09:42:02.000000000","tz":120},"subject":"Update rootfs.tar.xz for kuryr/demo","message":"Update rootfs.tar.xz for kuryr/demo\n\nSeems like rootfs.tar.xz was not updated after upd_client was added to\nkuryr/demo container image. This commit updates it.\n\nChange-Id: Id0fa12c3bb240d1d8eae01db78558e7f908f80a9\nRelated-Bug: 1799239\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/44fa61591d545ea8eaf3f1e6be1bed9797ebca8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/44fa61591d545ea8eaf3f1e6be1bed9797ebca8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
