)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~663090","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I6522689fac6a7398cd78665a13daa6f5e46c6be2","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"remove-multithread","hashtags":[],"change_id":"I6522689fac6a7398cd78665a13daa6f5e46c6be2","subject":"Removing threading option","status":"MERGED","created":"2019-06-04 16:31:29.000000000","updated":"2019-06-19 16:54:07.000000000","submitted":"2019-06-19 16:54:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663090-1560963247833-eb0806c3","meta_rev_id":"ec120cfbb90ea29f3c15920a0faf758de88f53de","_number":663090,"virtual_id_number":663090,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":2,"date":"2019-06-19 16:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-18 10:17:23.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"all":[{"value":2,"date":"2019-06-19 14:58:15.000000000","_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":2,"date":"2019-06-19 12:26:42.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"all":[{"value":1,"date":"2019-06-19 14:58:15.000000000","_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 16:32:19.000000000","updated_by":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"reviewer":{"_account_id":21302,"name":"Gena","email":"gcheresh@redhat.com","username":"gcheresh"},"state":"REVIEWER"},{"updated":"2019-06-05 06:34:03.000000000","updated_by":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"reviewer":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"state":"REVIEWER"},{"updated":"2019-06-19 12:26:42.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-06-19 14:58:15.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2019-06-19 16:54:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1644ff4664577c3f6902ce76607e515ddde6daf","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-04 16:31:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f386b657061053700858bc486340c795f34007c","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-04 16:32:10.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"1a4ab8d631097eec6e1a73971502c7e6f697f0a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 19:03:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/90/663090/1/check/openstack-tox-docs/dd57a6d/html/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/90/663090/1/check/openstack-tox-pep8/bf7cd04/ : FAILURE in 4m 59s\n- kuryr-kubernetes-tempest http://logs.openstack.org/90/663090/1/check/kuryr-kubernetes-tempest/bf54cd6/ : SUCCESS in 1h 27m 48s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/90/663090/1/check/kuryr-kubernetes-tempest-openshift/ef943a4/ : FAILURE in 1h 58m 15s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/90/663090/1/check/kuryr-kubernetes-tempest-containerized/6c1e93b/ : SUCCESS in 1h 23m 29s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/90/663090/1/check/kuryr-kubernetes-tempest-containerized-ovn/a0d0bfa/ : FAILURE in 1h 33m 08s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/90/663090/1/check/kuryr-kubernetes-tempest-py36/ea2711b/ : SUCCESS in 1h 26m 54s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/90/663090/1/check/kuryr-kubernetes-tempest-containerized-py36/4ce0eae/ : SUCCESS in 1h 30m 25s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/90/663090/1/check/kuryr-kubernetes-tempest-multinode-containerized/f8413a5/ : FAILURE in 59m 49s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/90/663090/1/check/kuryr-kubernetes-tempest-centos-7/686d34c/ : SUCCESS in 1h 48m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac097888e37880bd05d055f60dd133f89995e148","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 19:51:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/90/663090/1/experimental/kuryr-kubernetes-tempest-containerized-openshift/203a740/ : TIMED_OUT in 3h 10m 33s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/90/663090/1/experimental/kuryr-kubernetes-tempest-containerized-l2/e066f13/ : FAILURE in 1h 35m 58s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/90/663090/1/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/1c9a5e7/ : TIMED_OUT in 3h 04m 42s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/90/663090/1/experimental/kuryr-kubernetes-tempest-containerized-network-policy/93f64b3/ : SUCCESS in 58m 39s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/90/663090/1/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/7d800ea/ : TIMED_OUT in 3h 04m 44s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/90/663090/1/experimental/kuryr-kubernetes-tempest-ovn/e88db44/ : FAILURE in 1h 33m 44s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/90/663090/1/experimental/kuryr-kubernetes-tempest-openshift-ingress/2e4fd26/ : FAILURE in 1h 02m 21s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/90/663090/1/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/515883d/ : FAILURE in 1h 19m 24s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/90/663090/1/experimental/kuryr-kubernetes-tempest-multinode-ha/e8f74b7/ : TIMED_OUT in 2h 16m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/90/663090/1/experimental/kuryr-kubernetes-tempest-containerized-crio/4434b8f/ : FAILURE in 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab0d66c2ce84def3a685004aad79c16f850928c2","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-05 06:33:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5077f587307b1d00ff52fc62a280b8bd13e0b63","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-05 06:33:53.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"005bcdac9a667c45110137adf7fbe7af426ac30f","author":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"date":"2019-06-05 07:29:19.000000000","message":"Patch Set 2: Code-Review-1\n\nDo we know any cause?","accounts_in_message":[],"_revision_number":2},{"id":"8c188898bb018959bbffcd5786f02d9918c8418d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 08:31:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/90/663090/2/check/openstack-tox-docs/fcfb8d6/html/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/90/663090/2/check/openstack-tox-pep8/98d3947/ : SUCCESS in 4m 56s\n- kuryr-kubernetes-tempest http://logs.openstack.org/90/663090/2/check/kuryr-kubernetes-tempest/8188eb8/ : SUCCESS in 1h 40m 01s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/90/663090/2/check/kuryr-kubernetes-tempest-openshift/18142a8/ : SUCCESS in 1h 53m 17s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/90/663090/2/check/kuryr-kubernetes-tempest-containerized/3936cba/ : SUCCESS in 1h 17m 22s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/90/663090/2/check/kuryr-kubernetes-tempest-containerized-ovn/61cb2a2/ : FAILURE in 1h 51m 40s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/90/663090/2/check/kuryr-kubernetes-tempest-py36/2baa360/ : SUCCESS in 1h 20m 15s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/90/663090/2/check/kuryr-kubernetes-tempest-containerized-py36/2ce4855/ : SUCCESS in 1h 41m 26s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/90/663090/2/check/kuryr-kubernetes-tempest-multinode-containerized/0d74d37/ : SUCCESS in 1h 40m 23s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/90/663090/2/check/kuryr-kubernetes-tempest-centos-7/efc123d/ : SUCCESS in 1h 39m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18831dbe01d442a721120f246be3519a774a4678","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-05 09:10:13.000000000","message":"Patch Set 2:\n\n\u003e Do we know any cause?\n\nyes! it creates many threads and running it locally I saw some errors related to not being able to create new threads, failing in cleanup tasks, as well as for some of the threads that the kuryr controller is creating\nFile \"/usr/lib64/python2.7/threading.py\", line747, in start\n _start_new_thread(self.__bootstrap, ())\nthread.error: can\u0027t start new thread","accounts_in_message":[],"_revision_number":2},{"id":"0577521a921305b25da90497499de916bda2c991","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 09:45:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-containerized-openshift/addb9f8/ : FAILURE in 1h 42m 40s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-containerized-l2/1d3601d/ : FAILURE in 1h 58m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/4f5b60e/ : TIMED_OUT in 3h 04m 23s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-containerized-network-policy/545e153/ : SUCCESS in 59m 50s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/a4d4b0d/ : SUCCESS in 2h 38m 45s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-ovn/06ed08a/ : FAILURE in 1h 29m 29s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-openshift-ingress/a87e27f/ : FAILURE in 1h 02m 08s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/0ffd79a/ : FAILURE in 1h 12m 52s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-multinode-ha/9ea5f71/ : TIMED_OUT in 2h 15m 36s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-containerized-crio/8df5855/ : FAILURE in 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c8187167c0428de42ad165e17d3c2aa1af66e0a","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-05 12:28:30.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"f81aa8bd7cf4d5042749eeaae4f6b0122e86f22a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 15:41:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kuryr-kubernetes-tempest-containerized-openshift http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-containerized-openshift/a625b74/ : FAILURE in 37m 49s (non-voting)\n- kuryr-kubernetes-tempest-containerized-l2 http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-containerized-l2/f9122df/ : FAILURE in 1h 31m 08s (non-voting)\n- kuryr-kubernetes-tempest-containerized-pools-namespace http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-containerized-pools-namespace/5015630/ : TIMED_OUT in 3h 05m 06s (non-voting)\n- kuryr-kubernetes-tempest-containerized-network-policy http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-containerized-network-policy/fd25264/ : SUCCESS in 1h 03m 48s (non-voting)\n- kuryr-kubernetes-tempest-containerized-openshift-serial http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-containerized-openshift-serial/83e8297/ : FAILURE in 2h 56m 32s (non-voting)\n- kuryr-kubernetes-tempest-ovn http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-ovn/a6212a1/ : FAILURE in 1h 30m 02s (non-voting)\n- kuryr-kubernetes-tempest-openshift-ingress http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-openshift-ingress/fb4f7f5/ : FAILURE in 1h 02m 25s (non-voting)\n- kuryr-kubernetes-tempest-openshift-multi-vif http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-openshift-multi-vif/f077266/ : FAILURE in 1h 26m 16s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ha http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-multinode-ha/bf160eb/ : FAILURE in 47m 34s (non-voting)\n- kuryr-kubernetes-tempest-containerized-crio http://logs.openstack.org/90/663090/2/experimental/kuryr-kubernetes-tempest-containerized-crio/3817c8a/ : FAILURE in 27m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab774cbee7724231173afdbab7d95e010003c09c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-06-05 15:51:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"56b552e2a6b6114bd43510dd1043b080da0ebfae","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-13 07:09:40.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3412ce66196db5f410c86b17e3d387475a8fa4f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:17:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/90/663090/3/check/openstack-tox-docs/69b00b4/html/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/90/663090/3/check/openstack-tox-pep8/0852688/ : SUCCESS in 4m 17s\n- kuryr-kubernetes-tempest http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest/d881d02/ : SUCCESS in 1h 19m 19s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-openshift/3b39209/ : FAILURE in 1h 57m 00s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-containerized/5bb732f/ : SUCCESS in 1h 23m 30s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-containerized-ovn/94eba2e/ : SUCCESS in 1h 27m 57s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-py36/604b7d9/ : SUCCESS in 1h 27m 26s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-containerized-py36/5ac33bd/ : SUCCESS in 1h 43m 35s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-multinode-containerized/b49f108/ : SUCCESS in 2h 00m 26s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-centos-7/4fa9a8e/ : SUCCESS in 1h 36m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ce4907b56e3a1aa9577f142d36c561e851f6493e","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-06-18 10:17:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"69dc7cc39e82078d9c62af06b0ebcef2b8957871","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 12:20:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/90/663090/3/check/openstack-tox-docs/8d529cb/html/ : SUCCESS in 6m 32s\n- openstack-tox-pep8 http://logs.openstack.org/90/663090/3/check/openstack-tox-pep8/b227505/ : SUCCESS in 4m 49s\n- kuryr-kubernetes-tempest http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest/8814aa1/ : SUCCESS in 1h 21m 40s\n- kuryr-kubernetes-tempest-openshift http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-openshift/a9aa56f/ : FAILURE in 1h 56m 02s (non-voting)\n- kuryr-kubernetes-tempest-containerized http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-containerized/3414226/ : SUCCESS in 1h 41m 19s (non-voting)\n- kuryr-kubernetes-tempest-containerized-ovn http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-containerized-ovn/f6ffe58/ : SUCCESS in 1h 27m 12s (non-voting)\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-py36/1070510/ : SUCCESS in 1h 25m 00s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-containerized-py36/8e88ecd/ : SUCCESS in 1h 27m 46s\n- kuryr-kubernetes-tempest-multinode-containerized http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-multinode-containerized/02b7c33/ : SUCCESS in 1h 49m 10s (non-voting)\n- kuryr-kubernetes-tempest-centos-7 http://logs.openstack.org/90/663090/3/check/kuryr-kubernetes-tempest-centos-7/65a9a40/ : SUCCESS in 1h 47m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"673d5354c5488917f816c1d9094131c53e93847b","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-06-19 12:26:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b430ce9e468dd6fed48e6c7d51308330897ea8b8","author":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"date":"2019-06-19 14:58:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bd5da079f8049489d4513d435db8c9498d5cc178","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 14:58:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bda56da5a21880d0909f05dc9615d57c5ea44676","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 16:54:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2cacf6f25ae0e8815d3380d95bc32a5362358051","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 16:54:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/90/663090/3/gate/openstack-tox-docs/4abbe8d/html/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 http://logs.openstack.org/90/663090/3/gate/openstack-tox-pep8/b234a5b/ : SUCCESS in 4m 18s\n- kuryr-kubernetes-tempest http://logs.openstack.org/90/663090/3/gate/kuryr-kubernetes-tempest/770ea42/ : SUCCESS in 1h 19m 18s\n- kuryr-kubernetes-tempest-py36 http://logs.openstack.org/90/663090/3/gate/kuryr-kubernetes-tempest-py36/5fa5d98/ : SUCCESS in 1h 34m 02s\n- kuryr-kubernetes-tempest-containerized-py36 http://logs.openstack.org/90/663090/3/gate/kuryr-kubernetes-tempest-containerized-py36/f0b22cc/ : SUCCESS in 1h 21m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"11463dda9e4feb26eac938aee31d203aea1e8675","revisions":{"a6d3acc2a323793b06fb3b134b4185ed86a1953b":{"kind":"REWORK","_number":1,"created":"2019-06-04 16:31:29.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/90/663090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/90/663090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/90/663090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/90/663090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/90/663090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/90/663090/1"}}},"commit":{"parents":[{"commit":"6c4ac2bebd235264f0787006d3478032dee450ea","subject":"detect failed curl when streamed from pod","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6c4ac2bebd235264f0787006d3478032dee450ea"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-04 16:24:52.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-04 16:24:52.000000000","tz":120},"subject":"Removing threading option","message":"Removing threading option\n\nGates are not reliably working with threading, leaving to some\nproblems and gates flakiness. Removing it to make it more stable\n\nChange-Id: I6522689fac6a7398cd78665a13daa6f5e46c6be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a6d3acc2a323793b06fb3b134b4185ed86a1953b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a6d3acc2a323793b06fb3b134b4185ed86a1953b"}]},"branch":"refs/heads/master"},"2b427231fc45ddb1465cd01a342186b9f42f2bff":{"kind":"REWORK","_number":2,"created":"2019-06-05 06:33:39.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/90/663090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/90/663090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/90/663090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/90/663090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/90/663090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/90/663090/2"}}},"commit":{"parents":[{"commit":"6c4ac2bebd235264f0787006d3478032dee450ea","subject":"detect failed curl when streamed from pod","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6c4ac2bebd235264f0787006d3478032dee450ea"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-04 16:24:52.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-05 06:33:02.000000000","tz":120},"subject":"Removing threading option","message":"Removing threading option\n\nGates are not reliably working with threading, leaving to some\nproblems and gates flakiness. Removing it to make it more stable\n\nChange-Id: I6522689fac6a7398cd78665a13daa6f5e46c6be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2b427231fc45ddb1465cd01a342186b9f42f2bff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2b427231fc45ddb1465cd01a342186b9f42f2bff"}]},"branch":"refs/heads/master"},"11463dda9e4feb26eac938aee31d203aea1e8675":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-13 07:09:40.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/90/663090/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/90/663090/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/90/663090/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/90/663090/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/90/663090/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/90/663090/3"}}},"commit":{"parents":[{"commit":"6c4ac2bebd235264f0787006d3478032dee450ea","subject":"detect failed curl when streamed from pod","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/6c4ac2bebd235264f0787006d3478032dee450ea"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-04 16:24:52.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2019-06-13 07:07:32.000000000","tz":120},"subject":"Removing threading option","message":"Removing threading option\n\nGates are not reliably working with threading, leaving to some\nproblems and gates flakiness. Removing it to make it more stable\n\nDepends-On: I35f3be8fc16d71d0d2ac01c5451d398dd631f118\n\nChange-Id: I6522689fac6a7398cd78665a13daa6f5e46c6be2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/11463dda9e4feb26eac938aee31d203aea1e8675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/11463dda9e4feb26eac938aee31d203aea1e8675"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
